)]}'
{"id":"openstack%2Fzun-tempest-plugin~730538","triplet_id":"openstack%2Fzun-tempest-plugin~master~I3864f5eb05321345bfc3b2fa7d08654665a84219","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I3864f5eb05321345bfc3b2fa7d08654665a84219","subject":"Fix stable branch py2 job","status":"MERGED","created":"2020-05-25 01:29:51.000000000","updated":"2020-05-25 03:52:20.000000000","submitted":"2020-05-25 03:51:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730538-1590378666834-b12ba7e5","meta_rev_id":"3a2981ec80aa370759e9885ac2fb1678cebd71ce","_number":730538,"virtual_id_number":730538,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-25 03:51:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-05-25 02:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-05-25 02:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 03:51:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0e32e828b6b4f7f64661156d646dccc19103ff0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-25 01:29:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e6c74f25cf5c6b620f02da568f0b962bb680ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 02:30:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7936ceef298e45208dd7a1b6e8e5bf29 : SUCCESS in 2m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d9afdfb492452f9cae3dcdb0d58889 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b5d8c10c9441c79c11941505c16c43 : SUCCESS in 3m 12s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/3faf72acaa3447638489c62b68bbf805 : SUCCESS in 41m 49s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/1c6fa0583de6472b9f2fdadc97d18535 : SUCCESS in 41m 01s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/ec194fb3c7e641c8ba74e6d1db9a9462 : SUCCESS in 41m 12s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/fbf585e7c9d0464fbaeb0c5e8515babc : SUCCESS in 49m 01s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/f0f2b4ca411b4fdeb14fcddbb19dc651 : SUCCESS in 47m 59s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d8425d5bf0694b4ab1eaafdbc3dfc05b : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5fbf72ebcbde72fb58bce0ebcad4b3bc4850bac6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-25 02:57:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"201c455221bc539d1307600701483159ba3e6aaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 02:57:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f098f6781278e4922f82071afe9cdafbdcef681d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 03:51:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/45ab8d3a2c6d4be3b6130fd2a510d78d : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f3fe77a305a4e0a8e6cf9614ab62947 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be9533665c8241fa9c21857eba0d5851 : SUCCESS in 3m 43s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d630df89292d4fefb4314a00bf59b190 : SUCCESS in 51m 26s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/1323a4ff55634b06b6b7edade25332eb : SUCCESS in 48m 36s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/ee2a132ae60549ffbedfa7654e8f200d : SUCCESS in 51m 23s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/e53a508a01a74f2b83eb016eb98707ac : SUCCESS in 49m 29s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/6ad5707cb1884d0b89c47e8a150ced26 : SUCCESS in 42m 52s","accounts_in_message":[],"_revision_number":1},{"id":"119957dce5ad7d530a6382f0074d8b49b5637322","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 03:51:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3a2981ec80aa370759e9885ac2fb1678cebd71ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 03:52:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8db5ca2e03842688b204b9a724cfa6c : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3dc2289538fab168485e8d07581d3e2272e40c07","revisions":{"3dc2289538fab168485e8d07581d3e2272e40c07":{"kind":"REWORK","_number":1,"created":"2020-05-25 01:29:51.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/38/730538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/38/730538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/730538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/730538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/730538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/730538/1"}}},"commit":{"parents":[{"commit":"01d03bb71edd5661b233feddb9afd52be76e5949","subject":"Update \"python setup.py build_sphinx\" to build_sphinx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/01d03bb71edd5661b233feddb9afd52be76e5949"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-25 01:26:47.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-25 01:26:47.000000000","tz":0},"subject":"Fix stable branch py2 job","message":"Fix stable branch py2 job\n\nThe hacking version is not available at python2, which breaks\nstable/stein py2 job. This commit makes hacking py3 only.\n\nChange-Id: I3864f5eb05321345bfc3b2fa7d08654665a84219\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/3dc2289538fab168485e8d07581d3e2272e40c07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/3dc2289538fab168485e8d07581d3e2272e40c07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
