)]}'
{"id":"openstack%2Fzun-tempest-plugin~731426","triplet_id":"openstack%2Fzun-tempest-plugin~master~If8b11ab488e0418b344ebe0693ea7ba0ddeb6ce3","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"If8b11ab488e0418b344ebe0693ea7ba0ddeb6ce3","subject":"Add py38 package metadata","status":"MERGED","created":"2020-05-28 04:00:57.000000000","updated":"2020-06-07 19:24:13.000000000","submitted":"2020-06-07 19:23:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731426-1591557780597-1687bc1b","meta_rev_id":"53ec6513a1bd683c3f75fdfbc84df6ed7ab0b371","_number":731426,"virtual_id_number":731426,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-07 19:23:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-06-03 02:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-06-03 02:12:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 02:12:24.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-06-07 19:23:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c93ccd7f7495ef9fc2689f6ebfd6b7df250cda99","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2020-05-28 04:00:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82252761eeac19355cdea3425c84eecf2115889f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 05:04:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/76b1792899b846f4ab609aacdf5daa05 : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42037da3d6ce470ea2651dccb2350a9b : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53aa09ec47c44c099dc5955f64f1285b : SUCCESS in 2m 54s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a0c4c671b5034f90819f702d550ce0cb : SUCCESS in 50m 45s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/9e5641e300b6459ea235aca31c732136 : SUCCESS in 48m 19s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/90bb256d82d9416483b750755caaed4d : SUCCESS in 51m 40s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/595321f6443a430a8eb290b45fa4cc8e : SUCCESS in 47m 53s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/980674165b274fbe8339f8a703683050 : SUCCESS in 51m 02s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/53e4ce8c209f475dbd7edc37ce556020 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":1},{"id":"19d35e92f8a29c6e842d99d85cf992fecb1ef224","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-02 14:32:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c774851cffa8f435c22b78f11781f49131828d53","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2020-06-03 00:10:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c5cd68b0a171fa8a61ca28af73e83fa2725fde70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 01:04:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/49075d125864498a8e6740feda1e6c6f : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a8db1da627e41d8b0ebb699ba9ae996 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d10dc985be6437e88c50a5ce1b5a485 : SUCCESS in 2m 47s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9582de3bd40f4ec98a6c2b5151776a9c : SUCCESS in 39m 49s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/c24aaea97bb344c3aa2b3d263491f479 : SUCCESS in 49m 54s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/18b49b1375c041199962480af0c4a137 : SUCCESS in 45m 27s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/0e58731c48c543f2965c2482a5fd724d : SUCCESS in 42m 59s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/8012122a6c3d498cac33b3ad8e574815 : SUCCESS in 45m 21s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/bd350bb2a69940b0b6f4dca6380eef1d : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c71d1601d04f45b4dd9c29beb5cdf9c29d08e0a6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-03 02:12:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7e4347545c9e7661761664d323f1a17138632e0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 02:12:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36a9070671ec8823161099287ec969a71c3e7846","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 03:16:58.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9682ba7696864e44be492c2dea900baf : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333181e64a004421abcad0f2158d6630 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6de923a4a9f14e1d9ca7625cf64a0847 : SUCCESS in 3m 11s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2c92c5bf0e0f44eeac61dfe38cb4e34b : FAILURE in 49m 20s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/3702af4155b24c92a8c01ae807cd3b36 : SUCCESS in 50m 02s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/f1249d6f11a34c4dbe57ad29cb9a2992 : SUCCESS in 48m 55s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/52d90ba304ae4dcaa32495150f4ac4bc : SUCCESS in 46m 54s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/f4ae1066b9a142fcbc9e07fc281dd8da : SUCCESS in 49m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8cf9e91b86b93ba20f766ce4806d01d0c8b7a16a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-07 17:13:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"52a54cbfcc2bdfe233ddda916aa676ccb0c26ad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 18:30:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/924c3ee018984b55b5747befb3a94d53 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73363eefc3e542bf89ecb83b0493f401 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e43d5e8d9b4bc79a4a06e700894643 : SUCCESS in 3m 22s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/0b72dc0977bd43b8a5382d09d26e8933 : SUCCESS in 46m 43s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/113700f3229647b7b6e13cd1961a8885 : SUCCESS in 47m 56s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/7469a9b297804549baf0b2e48b22d46f : SUCCESS in 43m 40s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/df7a625a159946f2b0ab04e47f0b11bf : SUCCESS in 44m 52s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/c51a3429d6c74394b239dd527a276bc2 : SUCCESS in 46m 55s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/8195762af3a24b359ee41c8b23d8675e : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":2},{"id":"e00f8031e934b6d6934a2b11f8d9c531e51a6217","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 18:30:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8560fd143138069a1f431eabbc609ffe7451e1d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 19:23:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"02088eca206175f0a768a0109b30adcbe3495f24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 19:23:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3149c02b9e6a4d56a2ae51b9cd765260 : SUCCESS in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86dbe5fd4b054363b3a563cf7f85ce3c : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee052a3f5d143aa910bb1ef54afa455 : SUCCESS in 2m 48s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/4f44930e602d4a7a9d62be340b289e8f : SUCCESS in 45m 48s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/56e9c0c9a7e24440a8f0a1981cb88f0b : SUCCESS in 33m 21s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/3944181ba75445728845f674583a6403 : SUCCESS in 34m 43s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/7b7ef2ca3a8d4494a131cb1a38050e0b : SUCCESS in 46m 40s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/685ccf86f73b45589b3d5004991a6b74 : SUCCESS in 34m 20s","accounts_in_message":[],"_revision_number":2},{"id":"53ec6513a1bd683c3f75fdfbc84df6ed7ab0b371","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-07 19:24:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d659667d4f4c42bc6213cac44bf17a : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"31366abe5c3fd689460657d074a476e6ca63883c","revisions":{"58ae4d37f994caa7daa1e1a888c08a15c8024b17":{"kind":"REWORK","_number":1,"created":"2020-05-28 04:00:57.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/26/731426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/26/731426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/26/731426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/26/731426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/26/731426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/26/731426/1"}}},"commit":{"parents":[{"commit":"3dc2289538fab168485e8d07581d3e2272e40c07","subject":"Fix stable branch py2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/3dc2289538fab168485e8d07581d3e2272e40c07"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2020-05-28 03:24:53.000000000","tz":480},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2020-05-28 04:00:47.000000000","tz":480},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nChange-Id: If8b11ab488e0418b344ebe0693ea7ba0ddeb6ce3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/58ae4d37f994caa7daa1e1a888c08a15c8024b17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/58ae4d37f994caa7daa1e1a888c08a15c8024b17"}]},"branch":"refs/heads/master"},"31366abe5c3fd689460657d074a476e6ca63883c":{"kind":"REWORK","_number":2,"created":"2020-06-03 00:10:21.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/26/731426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/26/731426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/26/731426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/26/731426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/26/731426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/26/731426/2"}}},"commit":{"parents":[{"commit":"3dc2289538fab168485e8d07581d3e2272e40c07","subject":"Fix stable branch py2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/3dc2289538fab168485e8d07581d3e2272e40c07"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2020-05-28 03:24:53.000000000","tz":480},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2020-06-03 00:10:03.000000000","tz":0},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nChange-Id: If8b11ab488e0418b344ebe0693ea7ba0ddeb6ce3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/31366abe5c3fd689460657d074a476e6ca63883c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/31366abe5c3fd689460657d074a476e6ca63883c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
