)]}'
{"id":"openstack%2Fzun-tempest-plugin~733996","triplet_id":"openstack%2Fzun-tempest-plugin~master~I89a25883b6341357d6d98b60445eba9d0fe69623","project":"openstack/zun-tempest-plugin","branch":"master","topic":"bug/1881910","hashtags":[],"change_id":"I89a25883b6341357d6d98b60445eba9d0fe69623","subject":"Test container in subnet without gateway_ip","status":"MERGED","created":"2020-06-07 23:38:51.000000000","updated":"2020-07-05 04:44:57.000000000","submitted":"2020-07-05 04:43:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733996-1593924231755-b4231495","meta_rev_id":"a4f4c9b7d78b97fbcbdfb29bb96b769897ae97c0","_number":733996,"virtual_id_number":733996,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-05 04:43:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-07-05 03:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-07-05 03:53:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-05 04:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e221549c0992bcb2349cd43ee85cb84ca8679bb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-07 23:38:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"277dbadede5f0519b4009412fbf103423000dd28","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-07 23:39:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d02e83e10dbd77d97f48cdc0648430e9a7cc65ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 00:41:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c898b781644e6d991c81df2ed4429e : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4e9fc18344947fab3d793be8ab4a151 : SUCCESS in 2m 52s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/4e402dfcc7ef48b2a33feabaef6552f3 : FAILURE in 43m 21s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/e5d01c156933412aa7e6d8d8294f609f : FAILURE in 48m 54s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/cedc9573940e445bb1fe81fb6d424760 : FAILURE in 46m 22s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/022358a007ec4399b386ba6a9a2a79a7 : FAILURE in 51m 03s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/776681d31c7e48be8201fa0d74fd657e : FAILURE in 37m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/6dc065a251b3408f887d920b003b81f9 : FAILURE in 1h 01m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3c9a687a446ea87071174b9ef14c562d0b3a7358","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-03 04:55:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4be4e9bf233eab7a387c86ae31d7ed5a3d2d2f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 06:09:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/146ce76c27cc4f19b492c7443fb5d566 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c02ebb8766d455b8b0c7ce0931d2482 : SUCCESS in 4m 09s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e610608cb91d4a5fac56ff745b0421f6 : SUCCESS in 51m 08s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/8b67e642148749c793c590001a3ca258 : FAILURE in 55m 16s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/ba6b455acde04ded9dd5e125f3fc001a : FAILURE in 53m 39s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/fd7c806b395d443a922fb2fde54e0b19 : SUCCESS in 51m 49s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/c5c7b532956b48d9952583cb638cb207 : SUCCESS in 1h 00m 07s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/22a09f5ba0a6426980d2b8a46410d262 : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":2},{"id":"5bbf625a4463c1461b0ee19868bc9b416633351c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-05 00:57:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ba448aa176b0088bbfde56e18cdd13905cdfb476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 01:59:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb3a2e641324c128fdadfb849e9b07c : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad0157d73a3b43ed8424027e60234521 : SUCCESS in 3m 08s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/92959ca8525c4e26b16f8e436b17e076 : SUCCESS in 48m 12s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/282643efdcee4c2cb0bd7e87dd0a7ae3 : SUCCESS in 46m 23s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/7288507506e0450b8759f018f4294115 : SUCCESS in 49m 03s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/f3e113ae79fd4b06bd9463c56225af8c : SUCCESS in 48m 10s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/10fe69a858b64d55a3f4359fb155f13f : SUCCESS in 49m 04s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/843f90ed177d441c836fcae0ac153ccb : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":2},{"id":"35a095f2422aadb40c180984f035d66d9584a9fc","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-05 03:53:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e63d5dfe6c646411790f281634284b2dc687622","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 03:53:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fdb03201d6e544f671c7d2459217f44f78746697","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 04:43:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"37099df86133376d9c59ee15c51d927b1a6d6a76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 04:43:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f3fd9557af4e328a641c5afe3ca09a : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b635f11c81f4d0098d646938b70b3e6 : SUCCESS in 3m 24s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d9d13d1bdc324c62bb7795279592c215 : SUCCESS in 47m 41s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/44281a71846546a79f11513392067632 : SUCCESS in 49m 07s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/5ab6e6c197eb412384c5105dae29ecc1 : SUCCESS in 46m 04s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/b252ca4bf7dd443a98daa153a333ba85 : SUCCESS in 47m 21s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/6d2b608826fe4b358f66912d563bbf36 : SUCCESS in 46m 55s","accounts_in_message":[],"_revision_number":2},{"id":"54d093808b41fef68b346611849ba78cedb335bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 04:44:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e151671af77460c9b14a64c19733425 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e6bcd4759a8c0ab20b8d51c3dec3a96b0a892f1","revisions":{"1df198e0bd0ddf5aec42b0b00ee782112b298ed5":{"kind":"REWORK","_number":1,"created":"2020-06-07 23:38:51.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/96/733996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/96/733996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/96/733996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/96/733996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/96/733996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/96/733996/1"}}},"commit":{"parents":[{"commit":"a02d0aa0fc79e4dee26fcaed7238bb19f461be74","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a02d0aa0fc79e4dee26fcaed7238bb19f461be74"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-07 23:36:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-07 23:38:50.000000000","tz":0},"subject":"Test container in subnet without gateway_ip","message":"Test container in subnet without gateway_ip\n\nChange-Id: I89a25883b6341357d6d98b60445eba9d0fe69623\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1df198e0bd0ddf5aec42b0b00ee782112b298ed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1df198e0bd0ddf5aec42b0b00ee782112b298ed5"}]},"branch":"refs/heads/master"},"1e6bcd4759a8c0ab20b8d51c3dec3a96b0a892f1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-07 23:39:14.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/96/733996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/96/733996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/96/733996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/96/733996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/96/733996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/96/733996/2"}}},"commit":{"parents":[{"commit":"a02d0aa0fc79e4dee26fcaed7238bb19f461be74","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a02d0aa0fc79e4dee26fcaed7238bb19f461be74"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-07 23:36:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-07 23:38:56.000000000","tz":0},"subject":"Test container in subnet without gateway_ip","message":"Test container in subnet without gateway_ip\n\nRelated-Bug: #1881910\nChange-Id: I89a25883b6341357d6d98b60445eba9d0fe69623\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1e6bcd4759a8c0ab20b8d51c3dec3a96b0a892f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1e6bcd4759a8c0ab20b8d51c3dec3a96b0a892f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
