)]}'
{"id":"openstack%2Fzun-tempest-plugin~745450","triplet_id":"openstack%2Fzun-tempest-plugin~master~I6d4f70976f74eb4167c033e1af1db6a0857fdd86","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I6d4f70976f74eb4167c033e1af1db6a0857fdd86","subject":"Remove assertion of neutron port tagging","status":"MERGED","created":"2020-08-09 05:32:41.000000000","updated":"2020-08-17 03:23:28.000000000","submitted":"2020-08-17 03:22:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745450-1597634546159-39e11680","meta_rev_id":"3cbeec0a730ae3140ef8953753f898c0b1b73908","_number":745450,"virtual_id_number":745450,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-17 03:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-08-17 02:20:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-08-17 02:20:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 03:22:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a511c295969b9e47a693c44375c390e098d0998","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-09 05:32:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"320fb5d154d73413a53d5b05afc1fa3bf53a152c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-09 06:36:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304dacd729ba490f843713e2633e9551 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2756be476f544266925611a64c819d01 : SUCCESS in 3m 09s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/469404ec8afd4ece897435561611aa9f : SUCCESS in 52m 30s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/2fb81258b7144b7d81d3019bfa6e9cce : SUCCESS in 50m 35s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/c36b0efd649b4dabaa29913e41ec55ec : SUCCESS in 48m 13s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/4bfc7d4ccd2a4753a3875218ed618f5f : SUCCESS in 46m 34s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/2d04f7d935f34a89a1b64275462a9f66 : SUCCESS in 47m 12s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/fd2aae1d4ef84838860a8b3e10656c32 : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":1},{"id":"38134fd91f8a667131c628f822cdfd511fecff8d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-08-17 02:20:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8eee6650d8a6e153292ac662767518b42b902636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 02:20:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ef8141286272e12545ea8a64ded78d4c38eca7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 03:22:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a269d2bbf048aea52f509d6553304b : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31442f2e529440bca9e8822bd2c1fda4 : SUCCESS in 2m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/36b98b701a3a4339898dd10796c4091b : SUCCESS in 50m 24s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/f8b8d4be28b74a21b869aa7a1ec40322 : SUCCESS in 54m 04s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/cf981dd71b3640f4a94b431a2c1462a4 : SUCCESS in 48m 29s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/96e20f53c7d74a42a8a2ef4c3e503d04 : SUCCESS in 47m 41s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/01428cbfb1e543fcaa1e3467d59f88ce : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4d49648a72854b95bfbbe8c4c344dea993417e71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 03:22:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3cbeec0a730ae3140ef8953753f898c0b1b73908","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 03:23:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d4342b96764bcbbefd4b64574b296c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5c6b0349e0c38eddf6ca84c7e857ea1e0f9c5d7","revisions":{"b5c6b0349e0c38eddf6ca84c7e857ea1e0f9c5d7":{"kind":"REWORK","_number":1,"created":"2020-08-09 05:32:41.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/50/745450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/50/745450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/50/745450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/50/745450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/50/745450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/50/745450/1"}}},"commit":{"parents":[{"commit":"4a69e89f382ebfbfc8e00251f8d90982e46c432a","subject":"Merge \"Test container in subnet without gateway_ip\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/4a69e89f382ebfbfc8e00251f8d90982e46c432a"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-09 05:31:09.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-08-09 05:32:40.000000000","tz":0},"subject":"Remove assertion of neutron port tagging","message":"Remove assertion of neutron port tagging\n\nNeutron port won\u0027t be tagging on CNI network driver.\n\nChange-Id: I6d4f70976f74eb4167c033e1af1db6a0857fdd86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b5c6b0349e0c38eddf6ca84c7e857ea1e0f9c5d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b5c6b0349e0c38eddf6ca84c7e857ea1e0f9c5d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
