)]}'
{"id":"openstack%2Fzun-tempest-plugin~750266","triplet_id":"openstack%2Fzun-tempest-plugin~master~I3d85fe16f9d992efb2365ec6c78f08323d5950d1","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I3d85fe16f9d992efb2365ec6c78f08323d5950d1","subject":"Use entrypoint in test_commit_container","status":"MERGED","created":"2020-09-08 01:28:36.000000000","updated":"2020-09-08 05:54:26.000000000","submitted":"2020-09-08 05:53:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750266-1599544393517-c6eda660","meta_rev_id":"424f522296c97c2a78d9661578827ef852a07c48","_number":750266,"virtual_id_number":750266,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 05:53:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-09-08 04:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-09-08 04:45:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 05:53:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1183ead6269c8382c4989566eb1d48211eae9f56","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-08 01:28:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efd754093999f13eac2d909a97c81f54e1efd417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 02:36:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac85c6474774e7a9368557592114c7e : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25691b05f33346729d29bafbe0201805 : SUCCESS in 2m 32s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b96e292934bd468f8ccf0536d052b694 : SUCCESS in 1h 01m 22s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/4fc16623a53b4a7f83efef1679cff340 : FAILURE in 1h 00m 24s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/3e16fd3cb83a4b569ca129bea9212b0b : FAILURE in 46m 09s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/006d40fd651845faa91f9935d7f502bd : FAILURE in 51m 09s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/f3e18bc21b44425085ae47ecc2b36858 : SUCCESS in 49m 20s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/eabaf63faf3748b98a92f87cb2f78708 : SUCCESS in 1h 02m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c63cb8f9803813908138ebcfc4f587e63a5479eb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-08 03:00:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d1a11b1b76aebac9dcba4351b772f90a6ab40f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 04:08:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48ccecd0153348ab81e0303cad07b03c : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1517dba9f00497584307e2914a8fdd7 : SUCCESS in 2m 28s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/e32f7573908d421fad73df4985a779ca : SUCCESS in 48m 19s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/7bf1dacb59274df6b291be14c36701d7 : SUCCESS in 52m 25s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/56698177b7814b20a7f285b7ab6cba4b : SUCCESS in 47m 12s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/1c799d1016c74f5bb2ae46ca0a783855 : SUCCESS in 47m 12s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/fe083c6d0b6249afb9e08ea48f14915b : SUCCESS in 50m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a7ebfddbb7884205aef230bffe9445a7 : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":2},{"id":"e44e5c979c6499d53fb03a11c37c74eaf2bb52d2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-08 04:45:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b591c5f9c2c531c27b0e28410818506485d8c3f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 04:45:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a13e9e81af7f2800d1085579e8da2d62453b6d36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 05:53:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ab2d2d03592497588067c6828586d02 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/530ce791ed7742bb8288280a6e5aad83 : SUCCESS in 2m 56s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2527e029e9cc468eaa3bb5079744bd68 : SUCCESS in 55m 15s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/41ecf1811ab44d41ba94d38262949b3f : SUCCESS in 53m 41s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/a048b891da364d04a7dc52730c9c8229 : SUCCESS in 51m 17s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/2cc7a91154cf432294f03bb84f967c18 : SUCCESS in 1h 00m 36s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/851a5d7d28224d43a07e0d9c28c62804 : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e13d414a7e465696342a0377516cdbee0e7b5ec3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 05:53:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"424f522296c97c2a78d9661578827ef852a07c48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 05:54:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc5d08f5b774004a9ff8b425ba5ba85 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6db3f1eb2fa89791dafa1c587ee8bdaa51f0386e","revisions":{"ce7198dfbb2d0051e4e7b9aa009e9a2b8ada83fc":{"kind":"REWORK","_number":1,"created":"2020-09-08 01:28:36.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/66/750266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/66/750266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/66/750266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/66/750266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/66/750266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/66/750266/1"}}},"commit":{"parents":[{"commit":"f88ab903185abd9d7915407be9c88dbde044a4ae","subject":"Merge \"Add container network connectivity test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f88ab903185abd9d7915407be9c88dbde044a4ae"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-08 01:24:13.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-08 01:24:13.000000000","tz":0},"subject":"Use entrypoint in test_commit_container","message":"Use entrypoint in test_commit_container\n\nMove \"/bin/sh -c\" from command to entrypoint which aligns with\nthe common practice.\n\nChange-Id: I3d85fe16f9d992efb2365ec6c78f08323d5950d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/ce7198dfbb2d0051e4e7b9aa009e9a2b8ada83fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/ce7198dfbb2d0051e4e7b9aa009e9a2b8ada83fc"}]},"branch":"refs/heads/master"},"6db3f1eb2fa89791dafa1c587ee8bdaa51f0386e":{"kind":"REWORK","_number":2,"created":"2020-09-08 03:00:43.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/66/750266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/66/750266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/66/750266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/66/750266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/66/750266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/66/750266/2"}}},"commit":{"parents":[{"commit":"f88ab903185abd9d7915407be9c88dbde044a4ae","subject":"Merge \"Add container network connectivity test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f88ab903185abd9d7915407be9c88dbde044a4ae"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-08 01:24:13.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-09-08 02:58:49.000000000","tz":0},"subject":"Use entrypoint in test_commit_container","message":"Use entrypoint in test_commit_container\n\nMove \"/bin/sh -c\" from command to entrypoint which aligns with\nthe common practice.\nSince entrypoint is supported since api version 1.40, we keep\nthe old test for testing api version earlier than 1.39.\n\nChange-Id: I3d85fe16f9d992efb2365ec6c78f08323d5950d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6db3f1eb2fa89791dafa1c587ee8bdaa51f0386e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6db3f1eb2fa89791dafa1c587ee8bdaa51f0386e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
