)]}'
{"id":"openstack%2Fzun-tempest-plugin~768529","triplet_id":"openstack%2Fzun-tempest-plugin~master~Idc05e0262fcfe09493516c45610f58177d6313d8","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"Idc05e0262fcfe09493516c45610f58177d6313d8","subject":"[DNM] Test CI","status":"ABANDONED","created":"2020-12-26 10:00:36.000000000","updated":"2020-12-29 10:42:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b242e6b73409a35e84d62f3f53599ee479070154","_number":768529,"virtual_id_number":768529,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-26 11:06:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-26 11:06:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"850cbf64e613d7b74c09007ce44b981065452294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-12-26 10:00:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10ccae13b2170057efc0450e1d9b15012a9a6a31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 11:06:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce41699baca649ed8e22796135ba8c63 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2962f4cd2258499395c5bf02c000ee28 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/384c9efb8db14d818b0d4f6db05c1f01 : SUCCESS in 4m 00s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/edc5bde32d6746aaabb0d66cb1dd77ce : SUCCESS in 55m 15s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/e7cd3d3e3c5c48b2946261265f1d97af : SUCCESS in 39m 06s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/f2ff39d12c574a14ac1c85bcf8becc8e : SUCCESS in 41m 09s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/5db35c1b93a0498085af460dcf920140 : FAILURE in 42m 23s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/29b70e35fd364e9fa6283eb84af304d0 : SUCCESS in 46m 02s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/dcf49ec012e0426cb44edc619c487125 : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":1},{"id":"335925807a2480eb00dc97feb01be032a647afe0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-12-26 11:46:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9e68e4280d0bfb2475cb011a87307a0adbf64cb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 13:01:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51416e103675415bb75bebd72d04efb4 : SUCCESS in 1m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee3e92595c340aeab086681a4eb3bd3 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b87575a787548be820417a590483f52 : SUCCESS in 2m 43s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b8748c223af04c15a3e8c87c22f11950 : SUCCESS in 47m 17s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/59f846c64a794a56af0f647cf63e7afc : SUCCESS in 47m 23s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/ee7ceada3bfd4ed2bb61a51170df381c : SUCCESS in 43m 31s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/fafbab5515bd4eb1a5e826cc7dfc30a6 : FAILURE in 43m 54s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/151da588f4774d0f804483f92997e940 : SUCCESS in 45m 34s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/197e266d281045b9b356a0770da4024b : SUCCESS in 1h 08m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b242e6b73409a35e84d62f3f53599ee479070154","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-12-29 10:42:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48dbecf285600f0f820c612c0b6499d13328e664","revisions":{"48dbecf285600f0f820c612c0b6499d13328e664":{"kind":"REWORK","_number":1,"created":"2020-12-26 10:00:36.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/29/768529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/29/768529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/29/768529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/29/768529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/29/768529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/29/768529/1"}}},"commit":{"parents":[{"commit":"9c28eca4f5bc4f4c300ed28994c33276913a5a84","subject":"Merge \"[goal] Migrate zun-tempest-plugin jobs to focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9c28eca4f5bc4f4c300ed28994c33276913a5a84"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 10:00:05.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-12-26 10:00:33.000000000","tz":480},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nChange-Id: Idc05e0262fcfe09493516c45610f58177d6313d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/48dbecf285600f0f820c612c0b6499d13328e664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/48dbecf285600f0f820c612c0b6499d13328e664"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
