)]}'
{"id":"openstack%2Fzun-tempest-plugin~795865","triplet_id":"openstack%2Fzun-tempest-plugin~master~Ib24d7b90a20f3b56993c1faede020c015395a1be","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ib24d7b90a20f3b56993c1faede020c015395a1be","subject":"Migrate from testr to stestr","status":"MERGED","created":"2021-06-10 18:42:46.000000000","updated":"2021-07-10 06:45:42.000000000","submitted":"2021-07-10 06:44:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"795865","meta_rev_id":"319e79e59e046885730d3c4791690b067f1c8d0d","_number":795865,"virtual_id_number":795865,"owner":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-10 06:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2021-07-10 05:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2021-07-07 11:34:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2021-07-10 05:37:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 18:52:33.000000000","updated_by":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-06-10 20:04:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-08 15:20:04.000000000","updated_by":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"74b26d6870d938e9892bc4301ad7f7d6feee0a88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-06-10 18:42:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b02118850c63d9e31ede7305dcf48bf874b9285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-06-10 19:12:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1903512c22fae32a81d8e34436281c359f603294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 20:04:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/95b19ce7ead34da2b2cb6e275e311b27 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab278533be394dffb3aef70f9bca7d7f : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad411dc5b3a4edda0cbc98cac6684ff : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cf6303d88b24f3b9d6388441748d4bc : FAILURE in 3m 32s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7d1909b812fc4f64a87cddcc3f0113a1 : FAILURE in 35m 32s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/fa4fca38694a42699e4af66be8a827d7 : SUCCESS in 47m 08s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/d4a17e2a962e4e488a554969a027f2e4 : SUCCESS in 45m 55s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/63ca28be179b420fb85b155531d2ff15 : SUCCESS in 45m 39s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/d3f5b634b17b4a5d8a56a78f97525342 : SUCCESS in 42m 47s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/094c3185333b4722995a0acce431c892 : FAILURE in 50m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c4cc14cffc96c730a66b5265dd3d980e3f0ea4d6","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-14 12:01:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nstestr and coverage dependencies are missing in test-requirements.txt \n\nHowever, it seems this plugin doesn\u0027t have any unit tests implemented - the tests in zun_tempest_plugin/tests are tempest tests , not unit ones - so coverage job is redundant here.","accounts_in_message":[],"_revision_number":2},{"id":"9601b39f680151e252286eadc3f7ff070baab674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-06-14 15:50:26.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b54ffd15bd0c3f2c49bfbcfdf84f19908b4bc99c","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-06-14 15:54:31.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5715aadf57fd866fd202e7da3ecc513c5f363534","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-06-14 15:56:53.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e (5 comments)\n\u003e \n\u003e stestr and coverage dependencies are missing in test-requirements.txt \n\u003e \n\u003e However, it seems this plugin doesn\u0027t have any unit tests implemented - the tests in zun_tempest_plugin/tests are tempest tests , not unit ones - so coverage job is redundant here.\n\nDoes that mean I should remove coverage jobs from .zuul.yaml right?","accounts_in_message":[],"_revision_number":3},{"id":"42d00a4d0b21e8cbbf9ac985c3905fbc2c5a1835","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-06-14 16:01:53.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 2: Code-Review-1\n\u003e \u003e \n\u003e \u003e (5 comments)\n\u003e \u003e \n\u003e \u003e stestr and coverage dependencies are missing in test-requirements.txt \n\u003e \u003e \n\u003e \u003e However, it seems this plugin doesn\u0027t have any unit tests implemented - the tests in zun_tempest_plugin/tests are tempest tests , not unit ones - so coverage job is redundant here.\n\u003e \n\u003e Does that mean I just add the stestr dependencies","accounts_in_message":[],"_revision_number":3},{"id":"0dd6ab814cc1664a177dbffdd2dec2166fcf3590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:44:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c1e8acddc33a44c4b93cedcd0d483843 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6220e2372ea245de848089488514fafe : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6e6bc5669d4121a556cf39fde95896 : SUCCESS in 4m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9acbd06a6d704affab246233158998be : FAILURE in 3m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/af0fe842b5fb45e4be64a673ed7c4000 : FAILURE in 35m 59s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/6b6b478b34e7487bbf1fc563bc39ce30 : SUCCESS in 50m 21s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/631f10fff5ea4950a68cee4fb1bef515 : SUCCESS in 43m 04s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/81e5863e7fd94bfd8e53d911300241f9 : FAILURE in 47m 41s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/ed758b546c7d4b609f02ba7098c3ac12 : SUCCESS in 49m 27s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b8d38996f7a846c4ac98e70d864ae148 : FAILURE in 47m 21s","accounts_in_message":[],"_revision_number":3},{"id":"c115a4cac5c78ec3dd97f43893f51825d157281a","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-16 09:18:30.000000000","message":"Patch Set 3:\n\n(1 comment)\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 2: Code-Review-1\n\u003e \u003e \n\u003e \u003e (5 comments)\n\u003e \u003e \n\u003e \u003e stestr and coverage dependencies are missing in test-requirements.txt \n\u003e \u003e \n\u003e \u003e However, it seems this plugin doesn\u0027t have any unit tests implemented - the tests in zun_tempest_plugin/tests are tempest tests , not unit ones - so coverage job is redundant here.\n\u003e \n\u003e Does that mean I should remove coverage jobs from .zuul.yaml right?\n\nyeah, I would say so","accounts_in_message":[],"_revision_number":3},{"id":"13e90f41d60362a5f4dbb821c9c8d7c17e906bf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-06-16 13:48:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec2383984609798ce7732819bbd781f9a53d9503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-06-16 13:54:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8c95e16de2ade1289188b9b6443e2d5e6fec8c19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 14:57:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/05343e349cf64f758601e53814479e9a : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1345d74b0b4ce0a800d0462b8aa54d : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de346401c8d401eb6dd8f05c9358e67 : SUCCESS in 4m 06s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/c3a36280803947f980020f9d26b5105c : FAILURE in 34m 03s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/9a7f988f3d98420f864ef45416587e9e : SUCCESS in 53m 33s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/9b511913611342e0b9f0ba375cbbca2d : SUCCESS in 54m 33s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/6e0216b4fdd34068b382dad5187e036e : SUCCESS in 45m 30s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/ea9a0a4bcdd74310bcf4750e5fb1dda5 : SUCCESS in 51m 12s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/87ba4de8e3724d4eb5e07c8b7e19eb8b : FAILURE in 53m 08s","accounts_in_message":[],"_revision_number":5},{"id":"f46441b3e31e75554f264962471c268ec27086ba","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-29 12:44:29.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"803c2a94cd465c30d5ef61f759a3781728449c17","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-06-29 18:20:16.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a1d58ead35721af9b6426e39744471318efa32c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-06-29 18:21:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1f727f9107e547145380bf80f9fecd23525bbc5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 19:25:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6f7c678793c1407c98ef818899288a38 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f8b4e7046144a1ab74ba19a8d62f4d : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d165a6f40adc4ad38ec97facc16a0c20 : SUCCESS in 5m 56s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2930277e1bb940dfb7ddbbe97df47d94 : FAILURE in 43m 39s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/7e3a2b61eb9c4c019bc04f58783d7b69 : SUCCESS in 54m 13s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/e7c31c9b41014973bb3daa87c479db6e : SUCCESS in 49m 12s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/b1a53ae71d874fd59b5f752ef7d69d72 : SUCCESS in 52m 46s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/9b7907383a024d21bd178172e46c8a2b : SUCCESS in 51m 58s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/8d9ada6ca48141bab921586dad8b6f30 : FAILURE in 58m 14s","accounts_in_message":[],"_revision_number":6},{"id":"698944b4a91d9a5b1f832481b242f78c76eb9eb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-07-01 10:39:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ac97dca6ea5c34f4d0ac80812d3f7f1035bcb031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 11:41:17.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a37151a5cf844b298fb37dc25b4a208b : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920a62d017ff40c684da61a4db5e31ac : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d6e240e24a4b4292461f5f4e555927 : SUCCESS in 3m 33s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2763fc0ba06a4060af071b70b3ee97c9 : FAILURE in 35m 23s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/5178b01ba81146028a7bcd3eb142ed70 : SUCCESS in 48m 56s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/e5fcb389743e43e4add554d3d9445a01 : SUCCESS in 45m 37s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/3df4add3f6e84c41a91415f8e867b362 : SUCCESS in 46m 41s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/b70e42c86b9c423c9459cbe0bda34ab3 : SUCCESS in 47m 57s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e9c9147749fe4a89ad6a5f52c15d6820 : FAILURE in 54m 10s","accounts_in_message":[],"_revision_number":7},{"id":"264e8adeee8efb3002574b1f86956ddba0dfc129","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-07-07 11:34:43.000000000","message":"Patch Set 7: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"03b20ff6711c2595aae2e7c79f706a976eb22f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 12:40:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/02b01c36599f4177968fcc05aae2b72f : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2ff061c35343c0b26af1889688e516 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/810c5a83a1a549c88928f67e5b15bdb5 : SUCCESS in 3m 45s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/af2813a302d74ee3b711841b963db8ef : SUCCESS in 43m 52s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/a38d359b3b4e4e24be749cdae111dfd3 : SUCCESS in 44m 30s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/6f1898e9a88941d4b8003f406402fbc4 : SUCCESS in 43m 49s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/17268990b23b4da68b4b3a5be46e381a : SUCCESS in 44m 38s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/7f898a3630954f2493f77f7bd3d86af3 : SUCCESS in 46m 56s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ba86d341ec8f46db9944c8ad7cc1f591 : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":7},{"id":"b81f8245b0b25423689f6fcd990e12788e1b6932","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-07-10 05:37:59.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"79e6a2747b7beb6d36bc663283b75e15962e12a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 05:38:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"13b5270e5bb14c1785f6333b6972a08b8d0e901b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 06:44:30.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/96a412a83e0046a2b71cc33abd278ca8 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2230b5322866435db809b5ecbaef961b : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24e7c13d7da4805a245d516c2d3f390 : SUCCESS in 4m 11s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/f6f745a75b204f81b717ae1e4f655804 : SUCCESS in 47m 51s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/e14c0325731e4b37b43b70586013fbed : SUCCESS in 1h 03m 44s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/a307bb9f8d0a409f953978cca8b64ddb : SUCCESS in 59m 51s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/33acda4b9d1e477ba77edd6c7db140b9 : SUCCESS in 47m 54s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/9bc233b1eeeb45e6b9fbc5b57dc3d03d : SUCCESS in 48m 48s","accounts_in_message":[],"_revision_number":7},{"id":"b01a3d080e3c0bda337a2bec8304750d3c7a0e28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 06:44:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"319e79e59e046885730d3c4791690b067f1c8d0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 06:45:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c32cbfafe54e7d816adf253eba9065 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"06283974c4aca968b09c196d81a5c06a8a930a49","revisions":{"e5fe7f57d7d334f1e59d20b5b4727eb50b52dd3e":{"kind":"REWORK","_number":1,"created":"2021-06-10 18:42:46.000000000","uploader":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"ref":"refs/changes/65/795865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/65/795865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/1"}}},"commit":{"parents":[{"commit":"749acf98d6b72a3a98823e8ed85c4a6e8929712c","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/749acf98d6b72a3a98823e8ed85c4a6e8929712c"}]}],"author":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-10 18:41:08.000000000","tz":330},"committer":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-10 18:41:08.000000000","tz":330},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore and test-requirements.txt file accordingly\n* Use py3 as the default runtime for tox\n* Add a new job, openstack-cover-jobs, to run the coverage in Zuul\n* Setuptools v54.1.0 introduces a warning that the use of dash-separated\n  options in \u0027setup.cfg\u0027 will not be supported in a future version [1].\n  Get ahead of the issue by replacing the dashes with underscores\n\n [1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: Ib24d7b90a20f3b56993c1faede020c015395a1be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/e5fe7f57d7d334f1e59d20b5b4727eb50b52dd3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/e5fe7f57d7d334f1e59d20b5b4727eb50b52dd3e"}]},"branch":"refs/heads/master"},"60d0630db04ab95e6a1cdddca3762575838bd6cd":{"kind":"REWORK","_number":2,"created":"2021-06-10 19:12:19.000000000","uploader":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"ref":"refs/changes/65/795865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/65/795865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/2"}}},"commit":{"parents":[{"commit":"749acf98d6b72a3a98823e8ed85c4a6e8929712c","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/749acf98d6b72a3a98823e8ed85c4a6e8929712c"}]}],"author":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-10 18:41:08.000000000","tz":330},"committer":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-10 19:11:44.000000000","tz":330},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore and test-requirements.txt file accordingly\n* Use py3 as the default runtime for tox\n* Add a new job, openstack-cover-jobs, to run the coverage in Zuul\n* Setuptools v54.1.0 introduces a warning that the use of dash-separated\n  options in \u0027setup.cfg\u0027 will not be supported in a future version [1].\n  Get ahead of the issue by replacing the dashes with underscores\n\n [1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: Ib24d7b90a20f3b56993c1faede020c015395a1be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/60d0630db04ab95e6a1cdddca3762575838bd6cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/60d0630db04ab95e6a1cdddca3762575838bd6cd"}]},"branch":"refs/heads/master"},"9ae0942d55198aa9771da2f5ef504a819b013a37":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-06-14 15:50:26.000000000","uploader":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"ref":"refs/changes/65/795865/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/65/795865/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/3"}}},"commit":{"parents":[{"commit":"749acf98d6b72a3a98823e8ed85c4a6e8929712c","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/749acf98d6b72a3a98823e8ed85c4a6e8929712c"}]}],"author":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-10 18:41:08.000000000","tz":330},"committer":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-14 15:50:26.000000000","tz":0},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore and test-requirements.txt file accordingly\n* Use py3 as the default runtime for tox\n* Add a new job, openstack-cover-jobs, to run the coverage in Zuul\n\nChange-Id: Ib24d7b90a20f3b56993c1faede020c015395a1be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9ae0942d55198aa9771da2f5ef504a819b013a37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9ae0942d55198aa9771da2f5ef504a819b013a37"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7d9f2f4452f28f2af8e756251d6703cb7ca746c4":{"kind":"REWORK","_number":4,"created":"2021-06-16 13:48:47.000000000","uploader":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"ref":"refs/changes/65/795865/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/65/795865/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/4"}}},"commit":{"parents":[{"commit":"749acf98d6b72a3a98823e8ed85c4a6e8929712c","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/749acf98d6b72a3a98823e8ed85c4a6e8929712c"}]}],"author":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-10 18:41:08.000000000","tz":330},"committer":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-16 13:46:25.000000000","tz":330},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore and test-requirements.txt file accordingly\n* Use py3 as the default runtime for tox\n\nChange-Id: Ib24d7b90a20f3b56993c1faede020c015395a1be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/7d9f2f4452f28f2af8e756251d6703cb7ca746c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/7d9f2f4452f28f2af8e756251d6703cb7ca746c4"}]},"branch":"refs/heads/master"},"5025ffd2ff9a656132a6ac20f6f007c0560979f5":{"kind":"REWORK","_number":5,"created":"2021-06-16 13:54:32.000000000","uploader":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"ref":"refs/changes/65/795865/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/65/795865/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/5"}}},"commit":{"parents":[{"commit":"749acf98d6b72a3a98823e8ed85c4a6e8929712c","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/749acf98d6b72a3a98823e8ed85c4a6e8929712c"}]}],"author":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-10 18:41:08.000000000","tz":330},"committer":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-16 13:53:47.000000000","tz":330},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore and test-requirements.txt file accordingly\n* Use py3 as the default runtime for tox\n\nChange-Id: Ib24d7b90a20f3b56993c1faede020c015395a1be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5025ffd2ff9a656132a6ac20f6f007c0560979f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5025ffd2ff9a656132a6ac20f6f007c0560979f5"}]},"branch":"refs/heads/master"},"1f9e3b75d7bd9873cb4c56c03bc2a3bccad8a561":{"kind":"REWORK","_number":6,"created":"2021-06-29 18:21:32.000000000","uploader":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"ref":"refs/changes/65/795865/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/65/795865/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/6"}}},"commit":{"parents":[{"commit":"749acf98d6b72a3a98823e8ed85c4a6e8929712c","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/749acf98d6b72a3a98823e8ed85c4a6e8929712c"}]}],"author":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-10 18:41:08.000000000","tz":330},"committer":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-29 18:20:58.000000000","tz":330},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore and test-requirements.txt file accordingly\n* Use py3 as the default runtime for tox\n\nChange-Id: Ib24d7b90a20f3b56993c1faede020c015395a1be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1f9e3b75d7bd9873cb4c56c03bc2a3bccad8a561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1f9e3b75d7bd9873cb4c56c03bc2a3bccad8a561"}]},"branch":"refs/heads/master"},"06283974c4aca968b09c196d81a5c06a8a930a49":{"kind":"REWORK","_number":7,"created":"2021-07-01 10:39:36.000000000","uploader":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"ref":"refs/changes/65/795865/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/65/795865/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/795865/7"}}},"commit":{"parents":[{"commit":"749acf98d6b72a3a98823e8ed85c4a6e8929712c","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/749acf98d6b72a3a98823e8ed85c4a6e8929712c"}]}],"author":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-06-10 18:41:08.000000000","tz":330},"committer":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-07-01 10:38:49.000000000","tz":330},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore and test-requirements.txt file accordingly\n* Use py3 as the default runtime for tox\n\nChange-Id: Ib24d7b90a20f3b56993c1faede020c015395a1be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/06283974c4aca968b09c196d81a5c06a8a930a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/06283974c4aca968b09c196d81a5c06a8a930a49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
