)]}'
{"id":"openstack%2Fzun-tempest-plugin~806492","triplet_id":"openstack%2Fzun-tempest-plugin~master~I219042307bea53ae56515c95b380b4c4cfb03c4d","project":"openstack/zun-tempest-plugin","branch":"master","topic":"bug/1941982","hashtags":[],"change_id":"I219042307bea53ae56515c95b380b4c4cfb03c4d","subject":"Temporarily disable capsule tests","status":"MERGED","created":"2021-08-29 08:29:45.000000000","updated":"2021-08-30 02:56:21.000000000","submitted":"2021-08-30 02:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806492-bug/1941982","meta_rev_id":"92c9e892b3ccdbca90e9275859ba840afdeef7a7","_number":806492,"virtual_id_number":806492,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-30 02:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2021-08-30 01:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2021-08-30 01:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-29 10:18:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1c976fb5bc319a8c9a1f5259f67f7894013e872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-08-29 08:29:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc405cb6e1ef4fa9a80397f6071b742e4711dbdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-08-29 09:19:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe73fe4b195d2e632829a90291bd6aa59c9367b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 10:18:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c6ab87c05649f192293a39e4d24e81 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98112add579341cb8a2db07cd5d0b406 : SUCCESS in 3m 58s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/4576949a1b7546228cafcc99ef13101e : SUCCESS in 35m 51s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/6b02f5db94224d7aba1791913b596996 : SUCCESS in 45m 00s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/b225f68f411d485889fd83ebae2ea369 : SUCCESS in 38m 57s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/1f60ea75681e47a9a0427ee0ca7bdb64 : SUCCESS in 42m 35s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/e20685fd657146c29ba2949d1e6a89e9 : SUCCESS in 42m 40s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a7d51cb28f134d27abae5fe19ad4d532 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":2},{"id":"741a73495af0d124cd694e5f711dcc84f7ab8508","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-08-30 01:54:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9009e9f4f5868887bd7ad91c413df16c0dbf5edc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 01:54:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"37a0bf63d9d4eb365e39cd7e8b3e9c903fc5fb92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 02:54:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d8897386d44d1baed81c4340095e5d : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd80b4a20a3f4825b9f9a8e71ec9bad8 : SUCCESS in 3m 51s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/4f63a42a81634a95908340d11db8c17c : SUCCESS in 39m 13s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/5e56d2c00afc400a860aea3950bc1470 : SUCCESS in 41m 59s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/dad9f0f5420a4fe183d5c2c1ed69b8f2 : SUCCESS in 49m 44s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/7082d31031d74895ad311f371b880c7b : SUCCESS in 54m 04s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/b3360673be374947905f9e05ac24262c : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":2},{"id":"10a4df0db81cfc1c1be4214bbf6e3fe30f85ed5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 02:55:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"92c9e892b3ccdbca90e9275859ba840afdeef7a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 02:56:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b6f3e3b92c4db6aaf40c3904eddc03 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d2529583bc32bcb84e447d5524af1c900cc2845","revisions":{"5519dd5904bfbd3a83f4b423c988383afaa51514":{"kind":"REWORK","_number":1,"created":"2021-08-29 08:29:45.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/92/806492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/92/806492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/806492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/806492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/806492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/806492/1"}}},"commit":{"parents":[{"commit":"9e0cf8fe919e6a807161b0e26363d532aa4cbdc8","subject":"Merge \"Migrate from testr to stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9e0cf8fe919e6a807161b0e26363d532aa4cbdc8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-08-29 08:28:51.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-08-29 08:29:43.000000000","tz":480},"subject":"Temporarily disable capsule tests","message":"Temporarily disable capsule tests\n\nRelated-Bug: #1941982\nChange-Id: I219042307bea53ae56515c95b380b4c4cfb03c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5519dd5904bfbd3a83f4b423c988383afaa51514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5519dd5904bfbd3a83f4b423c988383afaa51514"}]},"branch":"refs/heads/master"},"6d2529583bc32bcb84e447d5524af1c900cc2845":{"kind":"REWORK","_number":2,"created":"2021-08-29 09:19:40.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/92/806492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/92/806492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/806492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/806492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/806492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/806492/2"}}},"commit":{"parents":[{"commit":"9e0cf8fe919e6a807161b0e26363d532aa4cbdc8","subject":"Merge \"Migrate from testr to stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9e0cf8fe919e6a807161b0e26363d532aa4cbdc8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-08-29 08:28:51.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-08-29 09:19:30.000000000","tz":480},"subject":"Temporarily disable capsule tests","message":"Temporarily disable capsule tests\n\nRelated-Bug: #1941982\nChange-Id: I219042307bea53ae56515c95b380b4c4cfb03c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6d2529583bc32bcb84e447d5524af1c900cc2845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6d2529583bc32bcb84e447d5524af1c900cc2845"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
