)]}'
{"id":"openstack%2Fzun-tempest-plugin~811028","triplet_id":"openstack%2Fzun-tempest-plugin~master~I9a36e73480d271b6013c602ed02c8654dedf899c","project":"openstack/zun-tempest-plugin","branch":"master","attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-29 22:33:01.000000000","reason":"A robot voted negatively on a label"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-29 22:33:01.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9a36e73480d271b6013c602ed02c8654dedf899c","subject":"Revert \"Temporarily disable capsule tests\"","status":"NEW","created":"2021-09-26 09:01:42.000000000","updated":"2026-08-29 22:33:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"25c7bea68a975b925987daebffc0d0119a604a22","_number":811028,"virtual_id_number":811028,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-29 22:33:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-26 10:20:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 22:13:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"7aa7452fa7c32bda7a710d2ddb2a2abefd2d3320","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-09-26 09:01:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6034e72e273587a3cd456eeae786903c17be84af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 10:20:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9d774a965249ce8a4405b03703d4c6 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2888d76cfb814fb6974ff41d2385f8d5 : SUCCESS in 4m 30s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/75eee6b9199a441f928b62ddc7a2cc2c : FAILURE in 1h 06m 21s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/d5f864c39acd49309f1b277294acc028 : SUCCESS in 1h 11m 54s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/0b46f7ea465b4ae4a6d4703f82cc5f84 : SUCCESS in 45m 48s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/e2b0af99506948c492fde57e5c0fdc80 : SUCCESS in 43m 29s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/cf90ae0c89a94924a7f6f0b4b913dc80 : FAILURE in 52m 59s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/7897c78e97f447ba89dd13cea5ccdc5e : FAILURE in 1h 00m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d5b9bdfdc8a8664bc9e1a77fd42bc6c2f5f2835c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-10-04 04:18:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"61aa96182af0f3cb0d68c42f3f41e18a141bc714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 05:19:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ed15a09a3c4c6f9fb148ecdf61e5cc : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc85247b3e084ba78a1193db781fb73b : SUCCESS in 3m 43s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d3e418cfabcb416b82e0bf5a1c53f178 : FAILURE in 49m 16s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/b99f3cbb6af646f7a27538c773824514 : SUCCESS in 38m 23s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/e2ca59b8c6354a9b8c02a52a12f331db : FAILURE in 34m 19s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/5473d0db3f654ec99d4ab944a344387b : SUCCESS in 52m 10s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/42ac37d1f39649308e3a925580fb3c58 : FAILURE in 40m 37s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/99429c42698c4f6a8dd923eba6adfb5e : FAILURE in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a8a9ad0c5c7141c2f76625fc09afdc15c841d664","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-11-14 07:25:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a8051f3d6d9bf1195bdc97d2bacb2753c6c10b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 08:35:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6330013593a54fe19efb85ef9e317776 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2977683cfd8b4dcfbe5e31c83f6c7456 : SUCCESS in 3m 22s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/92a69c7827d9423d867608e9fa970f2b : FAILURE in 37m 47s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/1ac9cc21553e4e1683a7b3e8519f4bfc : SUCCESS in 47m 19s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/51ad68f3ad8d4d8894d3d8feaacc692e : SUCCESS in 47m 52s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/7d0867ecf0ec4344a7268c8eb7adddc1 : SUCCESS in 52m 56s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/705c01dd15b0405cb9aa9bb1c63601b5 : FAILURE in 53m 49s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f59b340378d741398f360cddecaf3b6e : FAILURE in 1h 04m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5be2eb9315bfa49da77693855f3cd63139a63a83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-11-14 10:15:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"269ff9b9eac0ddc809c35b8bbd6cf62c90a8b69d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-11-14 11:17:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2275fb03fbee939f491a44af3c1a6746521e545d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 12:17:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/addb1176551448dabc4c064984c63d52 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bbf7214870413db1027392bda7d2ec : SUCCESS in 3m 16s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7a6c25ff2bdc43b9be70978aa5b68f8d : FAILURE in 42m 58s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/f0882ea9524446f595b967301f9f07e9 : SUCCESS in 40m 09s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/0f066b939a4b4ebfb3668d2ffc4ee11b : SUCCESS in 43m 30s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/a93e012c64bc47838701faf758bf4d98 : SUCCESS in 32m 59s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/279605d4d07b47ccadfd9e8d62fb875d : FAILURE in 44m 41s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/92499f5b64d3406ca74d5946b782c831 : FAILURE in 58m 48s","accounts_in_message":[],"_revision_number":2},{"id":"565f17114bc2dccf6c393ce57c0661350cd0856d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-11-28 07:18:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d1add4ec3b0c84a13517576d57f3f7da09711ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-28 08:27:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efdc770c765941b7a2bdc60eedc77b26 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af8bb6867ad46b6bf38a357fdcf1555 : SUCCESS in 4m 29s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/04fc1496b13b4754bde58a7a3ee352ef : SUCCESS in 58m 42s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/aa3b7f28211a4a8c94e98f61d794a6a9 : SUCCESS in 41m 57s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/7dc79dee731a4b638c5c535e6603bead : SUCCESS in 52m 55s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/e49668b8c21f42edbb63ea0faac48c83 : SUCCESS in 1h 05m 20s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/9f3f143475714f71b280db0139f225f7 : FAILURE in 53m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/6efd761ce8f24fc0aba8a4a3de2ffbe4 : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":2},{"id":"ef14c3758ec44e12a366f503b8ace20d5bd2bb1e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-04-26 06:58:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed291a94c772292c09d3fadb444509eda30ebc16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 08:16:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa06b5e3b9004c3891e6d21c87e8431c : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f79370f2b3aa48789f7557ff9ed07ea5 : SUCCESS in 3m 15s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9a6f17535fdc47e78f3288560c47b497 : SUCCESS in 47m 07s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/cb7456d998a041678fe79c9adb91e43c : FAILURE in 19m 31s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/9f1f6d430dfa472ca77dbd385d734145 : TIMED_OUT in 1h 13m 40s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/853e9c06af064ca883c40d7512c20c89 : SUCCESS in 46m 49s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/6996def24bd64d0c822df8271d317242 : SUCCESS in 46m 16s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/0716e0f64fb64b599357776ad6bb0967 : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":2},{"id":"0039cbefc5cd9a016669ace19768e194923c10e7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-05-12 07:01:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94b22cf4f926b57bd6fe9a4e5724e5247e731bc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 08:06:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec907cfee893447bb4c30f735a9fd060 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fa2db8479f49b89afea4002a95c645 : SUCCESS in 3m 14s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/f2ddaaa8b5eb410a8ea6cfe6f235705b : SUCCESS in 50m 09s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/a280e8cebe00418b998e1574e5965bea : FAILURE in 46m 23s\n- zun-tempest-py3-docker-sql-victoria https://zuul.opendev.org/t/openstack/build/ddfe9eb8bb614fa7a76ed667f275de77 : FAILURE in 48m 45s\n- zun-tempest-py3-docker-sql-wallaby https://zuul.opendev.org/t/openstack/build/8a77ed07b1bc4536aaa9a53412fe4a7f : FAILURE in 51m 19s\n- zun-tempest-py3-docker-sql-xena https://zuul.opendev.org/t/openstack/build/b3100fc1c40e4c779b0416c3e33ffe4b : FAILURE in 45m 16s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/8f3a12bcd6f04714bbf13f3afa968948 : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":2},{"id":"c09e0e3f1ffa5f2bc2426fccda4915e94abe8bbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 14:29:37.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"02032d63fd9085d58825c9039cba66d70d8a152b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 14:29:56.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"331ce042b102480112fb4020242c4f502893ef8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 15:34:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c59a5b3e1f746fbb75f7bd766915df5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a467b1ee63145ba93f8bcffc5154eb0 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dd64939138349c4b7247c89e8e1492a : SUCCESS in 3m 10s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/dab66861ac2b4319806a126547773ca3 : FAILURE in 54m 45s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/d67b8fc77bcf4f5880d62bf097b189d1 : FAILURE in 48m 37s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/e9ddc15c62a54dd9ae196e64e7f13762 : FAILURE in 38m 50s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d2a193b8b4ec483d9bc3c4ea05e2c809 : FAILURE in 50m 32s","accounts_in_message":[],"_revision_number":4},{"id":"d76a26e4e1e076c11a05c09481765b382895889e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 16:27:59.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"14f30836c8ed31c3c2fb2a728e2b2915e6e280ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 17:31:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c2ec46a67e8485fa667759f9412ee86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a04b316bac47bf9009bfcc624238bf : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bafc325c7764ed8b735077348f08872 : SUCCESS in 4m 07s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/8342caf9686c48e8a105b4001c8e1c7b : FAILURE in 33m 25s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/33348a1f38d04bb7b49b96a75ffde7aa : FAILURE in 50m 37s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/dea7d98225644a109b91378c5dbc0901 : FAILURE in 30m 39s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/657636a463d74e63a31603fa7b5a41ca : FAILURE in 55m 19s","accounts_in_message":[],"_revision_number":5},{"id":"3ad51dce94579d1379883f08083924c13d1454ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 19:22:53.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"6e835e4f2df9318977a74bb8f6cdedc17e58395c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 20:21:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec1937d19b8547c7a4f0593bf7744217\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72edbbf4f3614ae8b8219e8e27da86ea : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42eced30822488fae114007f03f0b82 : SUCCESS in 2m 07s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/858c25b6b7314c8a9ffd015a28fd161a : FAILURE in 51m 11s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/aa9b598fb25e46e1b3a9a77c6688b182 : FAILURE in 41m 06s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/0b810e0f4bcc412a9e06b260c75cfbe1 : FAILURE in 37m 41s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/49bd736b9ed043c2b758dc27d8510f1f : FAILURE in 33m 40s","accounts_in_message":[],"_revision_number":6},{"id":"db4e755e6113c91bc818d3b80407cb3aef6ab443","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 22:13:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8fffab00e30a3a4debb14edb023a5d8039491a8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 21:31:15.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"25c7bea68a975b925987daebffc0d0119a604a22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 22:33:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/596fef377e4e442b88f8ef9ee7bf5f87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d5fb02ed4a4c7b99b405985ff892c5 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92dbc2c1644a41da855be70273c5c167 : SUCCESS in 2m 16s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/88eb2ec7c26a4144b436b90e9af2f9fc : FAILURE in 48m 49s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/d4ff7e39b06546f1aacd9c2fd6e133bb : FAILURE in 55m 30s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/8c85595969f5428e9e7c8bf3abe675a7 : FAILURE in 49m 11s\n- zun-tempest-py3-docker-sql-2026-1 https://zuul.opendev.org/t/openstack/build/27c436af7c7e4f67aabe9fc56cbb4166 : FAILURE in 47m 16s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/8b49af1d67c24615b5218ab808c8a23c : FAILURE in 39m 18s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4b40d60e20e14bb5f4d5aa921529d9902c06b34b","revisions":{"3b04ce13421912f72f86fef302c1555465633224":{"kind":"REWORK","_number":1,"created":"2021-09-26 09:01:42.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/811028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/28/811028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/1"}}},"commit":{"parents":[{"commit":"6d2529583bc32bcb84e447d5524af1c900cc2845","subject":"Temporarily disable capsule tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6d2529583bc32bcb84e447d5524af1c900cc2845"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-09-26 09:01:01.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-09-26 09:01:39.000000000","tz":480},"subject":"Revert \"Temporarily disable capsule tests\"","message":"Revert \"Temporarily disable capsule tests\"\n\nThis reverts commit 6d2529583bc32bcb84e447d5524af1c900cc2845.\n\nChange-Id: I9a36e73480d271b6013c602ed02c8654dedf899c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/3b04ce13421912f72f86fef302c1555465633224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/3b04ce13421912f72f86fef302c1555465633224"}]},"branch":"refs/heads/master"},"a78d9f411ee3a01881d40aad3289eecc19c2b29b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-14 10:15:33.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/28/811028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/28/811028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/2"}}},"commit":{"parents":[{"commit":"6d2529583bc32bcb84e447d5524af1c900cc2845","subject":"Temporarily disable capsule tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6d2529583bc32bcb84e447d5524af1c900cc2845"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-09-26 09:01:01.000000000","tz":480},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2021-11-14 10:15:33.000000000","tz":0},"subject":"Revert \"Temporarily disable capsule tests\"","message":"Revert \"Temporarily disable capsule tests\"\n\nThis reverts commit 6d2529583bc32bcb84e447d5524af1c900cc2845.\n\nDepends-On: https://review.opendev.org/c/openstack/zun/+/817868\nChange-Id: I9a36e73480d271b6013c602ed02c8654dedf899c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a78d9f411ee3a01881d40aad3289eecc19c2b29b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a78d9f411ee3a01881d40aad3289eecc19c2b29b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"87feacff4e1b0561286dcdf88e628886c8a0362d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-26 14:29:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/811028/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/28/811028/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/3"}}},"commit":{"parents":[{"commit":"6d2529583bc32bcb84e447d5524af1c900cc2845","subject":"Temporarily disable capsule tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6d2529583bc32bcb84e447d5524af1c900cc2845"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-09-26 09:01:01.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 14:29:37.000000000","tz":0},"subject":"Revert \"Temporarily disable capsule tests\"","message":"Revert \"Temporarily disable capsule tests\"\n\nThis reverts commit 6d2529583bc32bcb84e447d5524af1c900cc2845.\n\nDepends-On: https://review.opendev.org/c/openstack/zun/+/817868\nChange-Id: I9a36e73480d271b6013c602ed02c8654dedf899c\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/87feacff4e1b0561286dcdf88e628886c8a0362d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/87feacff4e1b0561286dcdf88e628886c8a0362d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"58ba044f934ec50ac7e6cba42a46bc3b24bd8e12":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-26 14:29:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/811028/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/28/811028/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/4"}}},"commit":{"parents":[{"commit":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","subject":"Remove EOl stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/60235d87783de40d2b80bbb2df3e2ea22e3b0fed"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-09-26 09:01:01.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 14:29:56.000000000","tz":0},"subject":"Revert \"Temporarily disable capsule tests\"","message":"Revert \"Temporarily disable capsule tests\"\n\nThis reverts commit 6d2529583bc32bcb84e447d5524af1c900cc2845.\n\nDepends-On: https://review.opendev.org/c/openstack/zun/+/817868\nChange-Id: I9a36e73480d271b6013c602ed02c8654dedf899c\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/58ba044f934ec50ac7e6cba42a46bc3b24bd8e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/58ba044f934ec50ac7e6cba42a46bc3b24bd8e12"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"6d2529583bc32bcb84e447d5524af1c900cc2845","ours":"87feacff4e1b0561286dcdf88e628886c8a0362d","theirs":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","merge_strategy":"recursive","contains_conflicts":false}},"900a5d101c2437a56bf909685b7af064e43ab3ff":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-06-26 16:27:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/811028/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/28/811028/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/5"}}},"commit":{"parents":[{"commit":"59699b11e8096304a8f095e84fa0b5b8da29834e","subject":"Bump docker api version to 1.40","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/59699b11e8096304a8f095e84fa0b5b8da29834e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-09-26 09:01:01.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 16:27:59.000000000","tz":0},"subject":"Revert \"Temporarily disable capsule tests\"","message":"Revert \"Temporarily disable capsule tests\"\n\nThis reverts commit 6d2529583bc32bcb84e447d5524af1c900cc2845.\n\nDepends-On: https://review.opendev.org/c/openstack/zun/+/817868\nChange-Id: I9a36e73480d271b6013c602ed02c8654dedf899c\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/900a5d101c2437a56bf909685b7af064e43ab3ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/900a5d101c2437a56bf909685b7af064e43ab3ff"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","ours":"58ba044f934ec50ac7e6cba42a46bc3b24bd8e12","theirs":"59699b11e8096304a8f095e84fa0b5b8da29834e","merge_strategy":"recursive","contains_conflicts":false}},"30f35f9f464438d3f6f7ef4c8100ac957ee2ef5d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-06-26 19:22:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/811028/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/28/811028/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/6"}}},"commit":{"parents":[{"commit":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a1e8e03c74a8ababfc61c419176fd06dfd7103ac"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-09-26 09:01:01.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 19:22:53.000000000","tz":0},"subject":"Revert \"Temporarily disable capsule tests\"","message":"Revert \"Temporarily disable capsule tests\"\n\nThis reverts commit 6d2529583bc32bcb84e447d5524af1c900cc2845.\n\nDepends-On: https://review.opendev.org/c/openstack/zun/+/817868\nChange-Id: I9a36e73480d271b6013c602ed02c8654dedf899c\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/30f35f9f464438d3f6f7ef4c8100ac957ee2ef5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/30f35f9f464438d3f6f7ef4c8100ac957ee2ef5d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"59699b11e8096304a8f095e84fa0b5b8da29834e","ours":"900a5d101c2437a56bf909685b7af064e43ab3ff","theirs":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","merge_strategy":"recursive","contains_conflicts":false}},"4b40d60e20e14bb5f4d5aa921529d9902c06b34b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-08-29 21:31:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/811028/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/28/811028/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/28/811028/7"}}},"commit":{"parents":[{"commit":"1c8845645f750337a7b9bf13a66af52562ada80a","subject":"Ensure that images for tempest tests configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-09-26 09:01:01.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:31:15.000000000","tz":0},"subject":"Revert \"Temporarily disable capsule tests\"","message":"Revert \"Temporarily disable capsule tests\"\n\nThis reverts commit 6d2529583bc32bcb84e447d5524af1c900cc2845.\n\nDepends-On: https://review.opendev.org/c/openstack/zun/+/817868\nChange-Id: I9a36e73480d271b6013c602ed02c8654dedf899c\nSigned-off-by: Hongbin Lu \u003chongbin034@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/4b40d60e20e14bb5f4d5aa921529d9902c06b34b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/4b40d60e20e14bb5f4d5aa921529d9902c06b34b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","ours":"30f35f9f464438d3f6f7ef4c8100ac957ee2ef5d","theirs":"1c8845645f750337a7b9bf13a66af52562ada80a","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
