)]}'
{"id":"openstack%2Fzun-tempest-plugin~838072","triplet_id":"openstack%2Fzun-tempest-plugin~master~I1297877db64364436d5160720bf33962024964e4","project":"openstack/zun-tempest-plugin","branch":"master","topic":"yoga-stable-job","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-12 12:56:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1297877db64364436d5160720bf33962024964e4","subject":"Add yoga stable branch jobs on master gate","status":"MERGED","created":"2022-04-15 02:16:20.000000000","updated":"2022-05-12 12:58:57.000000000","submitted":"2022-05-12 12:56:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838072-yoga-stable-job","meta_rev_id":"1ea54e393196e6fea0047c72f00faabf72a24b8f","_number":838072,"virtual_id_number":838072,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-12 12:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2022-05-12 12:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2022-05-12 12:07:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-15 04:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-26 06:58:32.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"},{"updated":"2022-05-12 12:07:31.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"7ebdc0a4a441bcf13234be0b2c61086e7fdbcafa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-15 02:16:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c374898033e7f43dd9be719e67369baae21a2e95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-15 02:18:16.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a079382a3e0980d88757108d9dd81d867bd0a165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 04:52:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623adde190f34074a3d38a1480f40b5f : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05860331af1b4d8fa1b1da4ede1d7ad8 : SUCCESS in 3m 28s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/de1bf79c48d8485ab600ce497e779b55 : TIMED_OUT in 1h 14m 01s\n- zun-tempest-py3-docker-sql-victoria https://zuul.opendev.org/t/openstack/build/aafe8cc0f4b0488eac8e7017e0c78b7f : FAILURE in 1h 01m 04s\n- zun-tempest-py3-docker-sql-wallaby https://zuul.opendev.org/t/openstack/build/5f084036e80a41cb9f61094d80b020f4 : SUCCESS in 48m 46s\n- zun-tempest-py3-docker-sql-xena https://zuul.opendev.org/t/openstack/build/c43383260ff3473883a0d6f3de6bf019 : SUCCESS in 47m 18s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/771d8387039746c294356ba6f3d0a842 : TIMED_OUT in 1h 14m 04s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/3053daa963594643869d8a754a9a2bce : TIMED_OUT in 2h 14m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4435cfd3af74cd6b8cc369ba207c605da417d553","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-04-26 06:58:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"850dab2f34b1308247a5de6ed7769d59dc1f4ee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 08:17:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70acf06ab9bf4d05a77f91f6183a07fe : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f585c2d0d764b5ca50cd746728fe39f : SUCCESS in 3m 07s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/bf6c5baa4436455bbf7d83608a70e6e7 : SUCCESS in 38m 55s\n- zun-tempest-py3-docker-sql-victoria https://zuul.opendev.org/t/openstack/build/c2f4f703a86f47029f711ec69b8cf445 : SUCCESS in 38m 32s\n- zun-tempest-py3-docker-sql-wallaby https://zuul.opendev.org/t/openstack/build/a4db1dc918164068bb65f072ed12223d : SUCCESS in 37m 40s\n- zun-tempest-py3-docker-sql-xena https://zuul.opendev.org/t/openstack/build/77b1d248819c44ada9e579153de06c45 : SUCCESS in 36m 36s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/d7c7902c6de443d2b4060bcad4e9ad84 : TIMED_OUT in 1h 13m 27s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/55eb3dfbb0f44dacb552c6cdbd4d2f32 : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":2},{"id":"26a0c80b67a6c5073e652db37e68921faad7c766","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-05-12 07:00:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"125a8ad1a3437dc10a318521374b0d4a10953b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 07:48:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ecc087630c040a08617313410764840 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67a2c3f606934a02bf3f60299e029a90 : SUCCESS in 2m 47s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/bfeb9ceb3cde4da6997b758a4335b835 : SUCCESS in 40m 19s\n- zun-tempest-py3-docker-sql-victoria https://zuul.opendev.org/t/openstack/build/ba9b0c5e465b4ae2b11c1f7ca8dafdd2 : SUCCESS in 39m 35s\n- zun-tempest-py3-docker-sql-wallaby https://zuul.opendev.org/t/openstack/build/a6cf0041c0544ae88c9cade0792e59e2 : SUCCESS in 42m 11s\n- zun-tempest-py3-docker-sql-xena https://zuul.opendev.org/t/openstack/build/2e480c8b015749afb7a1a771888664c3 : SUCCESS in 42m 15s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/7a8bb8b8357242a499368e5d8a092779 : SUCCESS in 45m 00s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/37b6177f7118401881f7b27ca06de6e4 : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":2},{"id":"99e6cdb8691d1da4c657c0a99cd500ee448d9057","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-05-12 12:07:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4918d9346391d6e29227ef894923861c6de25a6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 12:08:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"941d82eeeafb2e36bf60587603613cb1f467cbf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 12:56:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e52fb09bdfa460d9e1e906a3fee93ae : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dfabfdd2b784dbda71fcb4e763edf38 : SUCCESS in 3m 04s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7d1952ac59864ab3b0287f8916b741cf : SUCCESS in 44m 55s\n- zun-tempest-py3-docker-sql-victoria https://zuul.opendev.org/t/openstack/build/3815823822784947bef2bd7d26573264 : SUCCESS in 46m 31s\n- zun-tempest-py3-docker-sql-wallaby https://zuul.opendev.org/t/openstack/build/27bcc862a065447985a88f7b3f9ae73b : SUCCESS in 47m 47s\n- zun-tempest-py3-docker-sql-xena https://zuul.opendev.org/t/openstack/build/996324914c7241c5b85877e66ae537bf : SUCCESS in 42m 55s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/8be03b1a819345a7ac2261aab101e040 : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d92d4f87478a9202ec6a553fc6e35d43ada823b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 12:56:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1ea54e393196e6fea0047c72f00faabf72a24b8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 12:58:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cf5419767904441bcb00e8077812e7f : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c6b830885c15ed40e9e5436c88c72d32e754beae","revisions":{"0f21b772eccbb245c70cfa535ecd867da192a457":{"kind":"REWORK","_number":1,"created":"2022-04-15 02:16:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/72/838072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/72/838072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/72/838072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/72/838072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/72/838072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/72/838072/1"}}},"commit":{"parents":[{"commit":"538ba39e1484132a495c502e9e102135472cded9","subject":"Add victoria/wallaby/xena stable branch jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/538ba39e1484132a495c502e9e102135472cded9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-15 02:12:16.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-15 02:12:16.000000000","tz":-300},"subject":"Add wallaby|yoga stable branch jobs on master gate","message":"Add wallaby|yoga stable branch jobs on master gate\n\nWe have stable/yoga released now so we should add their\njob on master gate to keep branchless tempest plugins compatible\nto those branch.\n\nThis also removes the stable/ussuri job as it is in EM\nstate now.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I1297877db64364436d5160720bf33962024964e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/0f21b772eccbb245c70cfa535ecd867da192a457"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/0f21b772eccbb245c70cfa535ecd867da192a457"}]},"branch":"refs/heads/master"},"c6b830885c15ed40e9e5436c88c72d32e754beae":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-15 02:18:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/72/838072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/72/838072/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/72/838072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/72/838072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/72/838072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/72/838072/2"}}},"commit":{"parents":[{"commit":"538ba39e1484132a495c502e9e102135472cded9","subject":"Add victoria/wallaby/xena stable branch jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/538ba39e1484132a495c502e9e102135472cded9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-15 02:12:16.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-04-15 02:18:16.000000000","tz":0},"subject":"Add yoga stable branch jobs on master gate","message":"Add yoga stable branch jobs on master gate\n\nWe have stable/yoga released now so we should add their\njob on master gate to keep branchless tempest plugins compatible\nto those branch.\n\nThis also removes the stable/ussuri job as it is in EM\nstate now.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I1297877db64364436d5160720bf33962024964e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/c6b830885c15ed40e9e5436c88c72d32e754beae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/c6b830885c15ed40e9e5436c88c72d32e754beae"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
