)]}'
{"id":"openstack%2Fzun-tempest-plugin~844135","triplet_id":"openstack%2Fzun-tempest-plugin~master~I3fb61e06a31f0bb641909f45b9dda1ee1dfad5bc","project":"openstack/zun-tempest-plugin","branch":"master","topic":"victoria-pin-tempest","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2022-09-05 05:34:46.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-09-05 06:16:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3fb61e06a31f0bb641909f45b9dda1ee1dfad5bc","subject":"Remove stable/victoria job from master gate","status":"MERGED","created":"2022-05-31 21:08:36.000000000","updated":"2022-09-05 06:18:54.000000000","submitted":"2022-09-05 06:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844135-victoria-pin-tempest","meta_rev_id":"e82635d90fac8edba7b6d755357afd64b10b5704","_number":844135,"virtual_id_number":844135,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-05 06:16:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2022-09-05 05:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2022-09-05 05:34:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-31 22:11:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 15:40:14.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"8a145611c4d04576352f4eee5ad1e429feb49938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-31 21:08:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f1d61dc6d70be11a1f832aca30ef005ffe361c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 22:11:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0bab04fe66460da0becc06e0dc58c4 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ea3056e3e644315aee50086fa79afe0 : SUCCESS in 3m 34s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/1c4caf25791143ad8600b80b71a13f19 : SUCCESS in 42m 08s\n- zun-tempest-py3-docker-sql-wallaby https://zuul.opendev.org/t/openstack/build/b151dd18490747e59d750ad9e610d918 : SUCCESS in 46m 03s\n- zun-tempest-py3-docker-sql-xena https://zuul.opendev.org/t/openstack/build/e257a7acd6fa450a84cdf07b2e608c50 : SUCCESS in 41m 28s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/a20de40ea60e4be9bb297403a3708c44 : SUCCESS in 41m 13s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/edb387d4fd6642d3a0643bcef7fcda6c : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":1},{"id":"37022e908b1a0d4d188e21dbe35c174549e7a1a8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-09-05 05:34:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a00d9662cb102e7733826991b5be21b779225939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 05:35:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91ccfb171695428210bb2608144bf941db39ec96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 06:16:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d30bb67b8124cf6901922d794a42aad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360a4700dff44ba195d5e1b8b53c61df : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e621374ebcd54887be55446d182a403a : SUCCESS in 3m 24s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/47fc7b9bc7d04b4c87faf1c49cb18b0d : SUCCESS in 40m 37s\n- zun-tempest-py3-docker-sql-wallaby https://zuul.opendev.org/t/openstack/build/40397d3bbc9d49168578d0a6e875a6f2 : SUCCESS in 41m 13s\n- zun-tempest-py3-docker-sql-xena https://zuul.opendev.org/t/openstack/build/09d369478b444ad0b3e1da5c52fc0147 : SUCCESS in 34m 50s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/588b2846a00c488bb90c70037a6978aa : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6d2c12395b377e7f546578668867c55dda9d330a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 06:16:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e82635d90fac8edba7b6d755357afd64b10b5704","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 06:18:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a632bd38a9ba4fb29a6235c07d0327a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2132c015c4114cc789fc5a5abd8cc960 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8be4c7b08dd79da12d7c8962fb94c8f2ec4bc09","revisions":{"e8be4c7b08dd79da12d7c8962fb94c8f2ec4bc09":{"kind":"REWORK","_number":1,"created":"2022-05-31 21:08:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/844135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/35/844135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/35/844135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/35/844135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/35/844135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/35/844135/1"}}},"commit":{"parents":[{"commit":"c6b830885c15ed40e9e5436c88c72d32e754beae","subject":"Add yoga stable branch jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/c6b830885c15ed40e9e5436c88c72d32e754beae"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-31 21:05:14.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-31 21:05:14.000000000","tz":-300},"subject":"Remove stable/victoria job from master gate","message":"Remove stable/victoria job from master gate\n\nstable/victoria is in EM state and Tempest is pinned\nthere so we do not need to test the stable/victoria\njobs on master gate.\n\nChange-Id: I3fb61e06a31f0bb641909f45b9dda1ee1dfad5bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/e8be4c7b08dd79da12d7c8962fb94c8f2ec4bc09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/e8be4c7b08dd79da12d7c8962fb94c8f2ec4bc09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[]}
