)]}'
{"id":"openstack%2Fzun-tempest-plugin~885682","triplet_id":"openstack%2Fzun-tempest-plugin~master~I4558cc01c224ba9f57231c494270b396f6736f1e","project":"openstack/zun-tempest-plugin","branch":"master","topic":"2023-1-stable-job","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2023-06-22 12:17:15.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-06-22 14:36:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4558cc01c224ba9f57231c494270b396f6736f1e","subject":"Add stable/2023.1 jobs on master gate","status":"MERGED","created":"2023-06-09 04:11:51.000000000","updated":"2023-06-22 14:38:25.000000000","submitted":"2023-06-22 14:36:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885682-2023-1-stable-job","meta_rev_id":"2e5911432446d1a46937a711e3395fc563eb0d67","_number":885682,"virtual_id_number":885682,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-22 14:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2023-06-22 12:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2023-06-22 12:17:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-09 06:14:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-12 01:55:59.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"},{"updated":"2023-06-22 12:17:15.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"e104918a76bac0f7ac787d91db67f8bb6dafeb24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-06-09 04:11:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d200042ccad06aea72ed2b6a2c0d05d39ade206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-06-09 05:05:41.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"62cbe136917911d574869c3f0f349596cd6d3e0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 06:14:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d478cae916104a07894373aeff9648d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7729d3c35f0442ffb02b0cb18f01db3f : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719c0308b1a146ebbb5d84f3a496ad0b : SUCCESS in 2m 33s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/39288bdf115c4a688f7eef6e2e2da11a : SUCCESS in 48m 51s\n- zun-tempest-py3-docker-sql-xena https://zuul.opendev.org/t/openstack/build/8fb45f39a40842d991bf4e9d7e4c194b : FAILURE in 30m 43s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/03f9beceef0c4b5c9336b0d4fc3dcbc3 : FAILURE in 32m 23s\n- zun-tempest-py3-docker-sql-zed https://zuul.opendev.org/t/openstack/build/7e83e35b9f8d41538ae11cc58b308c1c : FAILURE in 38m 48s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/40e343a574b54d24991ade52ce25b2ed : SUCCESS in 50m 50s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a3b721d23023434c8dbcfe8c3929df94 : FAILURE in 1h 00m 15s","accounts_in_message":[],"_revision_number":2},{"id":"bab2b007316c6c4c0f71afa8cf82ef2f42c81b21","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-06-12 01:55:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8fcd3275851fbe998e61152432e47356e06d2d2a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-06-21 04:31:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce5753554621dedd679878b865b95f8f623fdcc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 05:27:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8d9287b14a94ff5bc97ca3b4958ceb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e50205496cf4dcf92ecf2b1e954714a : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5482e2bca420415aaa6e112632cd979c : SUCCESS in 3m 21s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d4fc203060d54811bb2e775755682d31 : SUCCESS in 43m 47s\n- zun-tempest-py3-docker-sql-xena https://zuul.opendev.org/t/openstack/build/f40264d2b00c4897863bf47de7daad83 : FAILURE in 24m 16s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/f12c641cfe5c4e2a8bf9af17263c472c : FAILURE in 27m 24s\n- zun-tempest-py3-docker-sql-zed https://zuul.opendev.org/t/openstack/build/bad01960f0ce49ca9a36debf5aa74171 : FAILURE in 25m 32s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/cf850e5f1ed74072ba838fdfe813be5b : SUCCESS in 44m 55s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/281cdbd111b74c6d9f20f205f94d8999 : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f11ba60bcbfa357b38a551abe54248b916006d23","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-06-21 14:44:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f0541211d67e01cc42fcd5781e132e1c2eb69be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:12:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5a508e6643b434da126bfded0b4a91a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e82bbde2f1ed4ac883c85761c225108d : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82c49dbca2b487b92ea009bb32d8020 : SUCCESS in 2m 52s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/c0f948f516c94f0eb892ad04030848a8 : TIMED_OUT in 1h 13m 20s\n- zun-tempest-py3-docker-sql-xena https://zuul.opendev.org/t/openstack/build/f62453c16ce64bf99d6d8c88832d7cab : FAILURE in 34m 06s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/ae82f2324db74bc8ba9e4a22c63fc57a : FAILURE in 26m 07s\n- zun-tempest-py3-docker-sql-zed https://zuul.opendev.org/t/openstack/build/bbeb30db5c7a4b37911a7d8a77db8f12 : TIMED_OUT in 1h 12m 39s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/9a78100cff3140b4b554155872569957 : SUCCESS in 1h 01m 26s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ba0d15ccf9ca41a2a83a5fe49861c816 : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":2},{"id":"58cd837bf3c423b54a79cc44ca5d55ffa1dcc879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-06-22 10:45:52.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"93085d8dd420f4ee522bc8e81e1b3efb282d4c78","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-06-22 10:46:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cad61106a1c55fdd1ecfee54ef261abb75e10e98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 11:48:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9046a9adf191407c8dbb3bd00a494674\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d41016bfe74a37a18f3186c67f614f : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bfdb25127ad4b1396bc4d38ecf95cc4 : SUCCESS in 2m 47s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/04c8f00737af48e2b37e5c7c7a7e2ed1 : SUCCESS in 44m 04s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/43237d1c317646488021aac4ec971cfd : FAILURE in 26m 50s\n- zun-tempest-py3-docker-sql-zed https://zuul.opendev.org/t/openstack/build/572cb9d91f4b4e84837264b1d02733bb : SUCCESS in 51m 01s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/cc0ca66558bc4a808470b699d4d573f0 : SUCCESS in 53m 53s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e05e17ae171f42a28bf96dc66264a9b9 : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":3},{"id":"767ef2964cedac33f7cb4546835d77ec8bda41df","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-06-22 12:17:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79b9b8dad563275103efe1adc2137c7666154e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 13:24:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad845433a6ff4d299faedc21deb7dbde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9961ee6df1d49598d41d6790a59fc3d : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47d906d5696a459586103269c9c3c4da : SUCCESS in 2m 45s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d8b7cc90d4ba45b484a9ed628bfd5f93 : SUCCESS in 48m 04s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/a9967d2acc6048cebb567e795a4c23f0 : SUCCESS in 56m 05s\n- zun-tempest-py3-docker-sql-zed https://zuul.opendev.org/t/openstack/build/635ec02fd1534ada8d5e7e22e0d1704d : SUCCESS in 49m 59s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/47ddc4f4a0344ea4904ca951dde33bc9 : SUCCESS in 50m 46s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/3701c627b61c4f748c7539d9295341fa : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":3},{"id":"bae81c3d04603b32e8276783491272b4838557aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 13:25:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5784e6ad22400e9ad2ae83561816dcbe93d50e8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 14:36:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61a1f8330b3a458090fdd0a9a57f0d67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2299b024a9bc4f46a3d4b3a7c3e84142 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c72b4b15b7734de3b3c5819c426076a7 : SUCCESS in 2m 37s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/cd4ecd61e7e7407e948ff90e52e3797e : SUCCESS in 54m 05s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/72cf746d94534c648461434f54861f1e : SUCCESS in 43m 40s\n- zun-tempest-py3-docker-sql-zed https://zuul.opendev.org/t/openstack/build/6565b8508845467385b8d65076c4b063 : SUCCESS in 59m 49s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/d4c475dccd4e44d38f3a2b6d38815526 : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":3},{"id":"87ba1ab4973e1683e8aefc86b388493a70641033","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 14:36:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2e5911432446d1a46937a711e3395fc563eb0d67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 14:38:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/140df0c4a1cd4bd59e6ff998eda5cf78\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03020b524cf74bbf95b1092a18f1d73b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1","revisions":{"b44e76460e2a441bf4ed4cc6589a3d6275b744da":{"kind":"REWORK","_number":1,"created":"2023-06-09 04:11:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/82/885682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/82/885682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/1"}}},"commit":{"parents":[{"commit":"29250bda508c95d0e24f383ca159016385326536","subject":"Merge \"Remove stable/wallaby job from master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/29250bda508c95d0e24f383ca159016385326536"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-06-09 02:15:24.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-06-09 02:15:24.000000000","tz":-300},"subject":"Add stable/zed jobs on master gate","message":"Add stable/zed jobs on master gate\n\nAs zed is released, we should add its job on master\ngate to keep branchless tempest plugins compatible\nto stable branch.\n\nAlso, remove stable/wallaby jobs given that wallaby went\nto Extended Maintenance on 2022-11-07T10:52:23+00:00\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I4558cc01c224ba9f57231c494270b396f6736f1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b44e76460e2a441bf4ed4cc6589a3d6275b744da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b44e76460e2a441bf4ed4cc6589a3d6275b744da"}]},"branch":"refs/heads/master"},"043a3388c110d774ae4df5f6043c88c58c8813d2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-06-09 05:05:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/82/885682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/82/885682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/2"}}},"commit":{"parents":[{"commit":"29250bda508c95d0e24f383ca159016385326536","subject":"Merge \"Remove stable/wallaby job from master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/29250bda508c95d0e24f383ca159016385326536"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-06-09 02:15:24.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-06-09 05:05:41.000000000","tz":0},"subject":"Add stable/2023.1 jobs on master gate","message":"Add stable/2023.1 jobs on master gate\n\nAs 2023.1 is released, we should add its job on master\ngate to keep branchless tempest plugins compatible\nto stable branch.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I4558cc01c224ba9f57231c494270b396f6736f1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/043a3388c110d774ae4df5f6043c88c58c8813d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/043a3388c110d774ae4df5f6043c88c58c8813d2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1":{"kind":"REWORK","_number":3,"created":"2023-06-22 10:45:52.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/82/885682/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/82/885682/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/82/885682/3"}}},"commit":{"parents":[{"commit":"29250bda508c95d0e24f383ca159016385326536","subject":"Merge \"Remove stable/wallaby job from master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/29250bda508c95d0e24f383ca159016385326536"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-06-09 02:15:24.000000000","tz":-300},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2023-06-22 10:45:44.000000000","tz":0},"subject":"Add stable/2023.1 jobs on master gate","message":"Add stable/2023.1 jobs on master gate\n\nAs 2023.1 is released, we should add its job on master\ngate to keep branchless tempest plugins compatible\nto stable branch.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I4558cc01c224ba9f57231c494270b396f6736f1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
