)]}'
{"id":"openstack%2Fzun-tempest-plugin~917074","triplet_id":"openstack%2Fzun-tempest-plugin~master~I9b2e2ba2336d285cf6ac35fb1ebeb78cf576a872","project":"openstack/zun-tempest-plugin","branch":"master","topic":"2023-1-last","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-12-06 04:35:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b2e2ba2336d285cf6ac35fb1ebeb78cf576a872","subject":"Update stable branch jobs on master gate","status":"MERGED","created":"2024-04-26 00:13:23.000000000","updated":"2024-12-06 04:37:34.000000000","submitted":"2024-12-06 04:35:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917074-2023-1-last","meta_rev_id":"0fb90696063d7141f8bdfcb7cb0f24476bef220e","_number":917074,"virtual_id_number":917074,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-06 04:35:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2024-12-06 02:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2024-12-06 02:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-26 01:02:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-04 05:44:03.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"},{"updated":"2024-12-06 02:57:39.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"b6d88b036ba610c04862edf37305e203ec467d88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-26 00:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6544f8e984f85b4ae6bb927d769fac3ed83757aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 01:02:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a514952d73704bf29531e0048a00667c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2891d6bdcfc4d21a7791e71015a2532 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa3268be419f4e3282cb42ecec6a049d : SUCCESS in 2m 40s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/81850edc1cf248fbb9c6c64bfc38c5c4 : FAILURE in 41m 04s\n- zun-tempest-py3-docker-sql-zed https://zuul.opendev.org/t/openstack/build/e7c6b28007d5462f822ed0810be4d464 : SUCCESS in 35m 38s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/1632ecab97944bc398e3ef1805781dad : SUCCESS in 34m 39s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/fc0b0ebf9de44aa085fb4e7ed1b8d33d : SUCCESS in 35m 03s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/dbd1570799fe4b66a38a6b3f9c8ef5c0 : FAILURE in 30m 19s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d205a09fe54f474bb8b9fc51997ebf67 : FAILURE in 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"990e2f4bfce239d2e40ea7021d58d74cc2b601d7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-05-04 05:44:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8d5501fe6ca10b6e4c1b636bd6ef12df9b1a4e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 06:39:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd80e004d36d4c11a3983c63dd2b6b44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8b64cec2f64349bf08d21d8bbd3a8c : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15a186a7d97b49928cb6db88b7d706a9 : SUCCESS in 3m 38s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/1a5afd63394b4c938c7e2c4831b84321 : SUCCESS in 38m 52s\n- zun-tempest-py3-docker-sql-zed https://zuul.opendev.org/t/openstack/build/678e60f6fa4c42e39bd204b4aa74bd49 : FAILURE in 31m 27s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/1d0d80ddbf0f490fb78eec3f1ba3eb55 : SUCCESS in 49m 12s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/06ecc34519c74f9ebb7d2a0ea4f44ec9 : SUCCESS in 28m 35s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/f748da6874254f26911535a284ed8962 : SUCCESS in 46m 40s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f696721330854b158b2f78664e7a5bc7 : SUCCESS in 30m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8b689938cd7597cea0c601b0ebfcd86d7b201c49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-06 17:50:23.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"63a0f64450c21012ba89f0e2d6722f8b53e7dfe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 18:48:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4eebeb6839064420bc7f9ca6a92af607\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4570766b70145d5a3c6baa81ea472e6 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f18cc29fefd4db589edbe3cee1b73ae : SUCCESS in 2m 28s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/5c11e124b56747538a0008e0b5e7bcdd : FAILURE in 27m 19s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/7870a5b424114c3bbe936556c290d319 : SUCCESS in 37m 58s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/f70def4486e0415d9c4532ad473d70e8 : SUCCESS in 36m 51s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/0a82c734cabd47ef8eb04f61be25ebe0 : SUCCESS in 57m 36s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/907d1e3d2b1641798dfa035c236852bf : FAILURE in 29m 02s","accounts_in_message":[],"_revision_number":2},{"id":"62587b2c9e705dc7cdd2e124724a3849a1d00848","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-10 15:46:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"138ba480a418c5069192a4ac2a90f84fa276520c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 16:25:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/794a7ed00540452a9583bc0a4746d26f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd36a1eb9ad442ab8fcf11e8be31aa4 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d9d0f5c9bfd433e9758373a1cb8cf55 : SUCCESS in 2m 20s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/91e84d5bb56646d2968625b95f286f88 : FAILURE in 16m 25s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/2f22fcfe17334f9089e437f949a7f81c : SUCCESS in 37m 41s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/3dc08f56239a4768a3f0a1a52c65737f : SUCCESS in 36m 45s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/94c0610c1aed43f38b239f54a8885c57 : SUCCESS in 28m 57s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a4b40a3a90a0435095561f3bd259f7dd : FAILURE in 17m 00s","accounts_in_message":[],"_revision_number":2},{"id":"265f40189c714d8bf15d5f38b3af60f5688ecf1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-24 18:37:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"309b4e11000ecdfb381b8cbdb6982bea479fa1ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:25:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cea10d9bba4548e69df713f26793954f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f18bcb9c94b4b788ccb30a3df9184b5 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7501fc2de6b47c58075d004430b7468 : SUCCESS in 2m 14s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/44ee264f14914cc6b32e4bfd5e8171bc : FAILURE in 27m 48s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/5513f6fc425746f9b6612dc3e774c9a1 : SUCCESS in 36m 17s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/1da45ff6fa9548319eba2093a42e9ce5 : SUCCESS in 31m 41s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/23630c3bb9a944f9b790dd4570a2de30 : SUCCESS in 33m 08s\n- zun-tempest-py3-docker-sql-2024-2 https://zuul.opendev.org/t/openstack/build/2356ea6fb7c94a1a815a4ab9da68abce : SUCCESS in 44m 04s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/fe9ee96afe0d488885f9b880d0dc1771 : FAILURE in 46m 18s","accounts_in_message":[],"_revision_number":3},{"id":"7bd9452ae083ec988bd8e7ecce616a61d4c325d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-06 19:52:45.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4575ab231c7339069e3ce5ead3699c0636980750","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-06 19:52:58.000000000","message":"Topic 2024-2-stable-job removed","accounts_in_message":[],"_revision_number":4},{"id":"a15735a5ffc3ac19534f199c166a0b0729323c8b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-06 19:53:05.000000000","message":"Topic set to 2023-1-last","accounts_in_message":[],"_revision_number":4},{"id":"35ec46ec3ecfc99c30acab506ca4afde286c94b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 20:53:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0dce90fb23f4ae3b9876e9f3b686a56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47835e72b5954077a5e02037391f2ae2 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fba08cabe291482692d5faf531a69dc1 : SUCCESS in 3m 06s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/0b24d6bd9cce4c30a5b2fbdeceae7a4b : FAILURE in 41m 29s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/bcc0daa11bb349cfae8992e5530b7281 : SUCCESS in 45m 15s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/8de7236aca8f469994371b81f76dd466 : SUCCESS in 47m 08s\n- zun-tempest-py3-docker-sql-2024-2 https://zuul.opendev.org/t/openstack/build/8149ab40cec3467982822f61fb92f795 : SUCCESS in 32m 11s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/b45ef8a50f864b8a8346ee887812b517 : FAILURE in 42m 00s","accounts_in_message":[],"_revision_number":4},{"id":"2cc77c4b97843c33087d00e37ba48775246aeaa2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-12-05 11:19:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"628d0718f70a1f1efd427f229b86f03eb3685ec4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-12-05 11:34:00.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"aa06cd52ee42f6e8267e124f772ba2fbe0bf8bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 12:28:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e18935bf7884fcda621b3edd0af5510\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cf2fbf0998c41db8b55bd40391cb885 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54838f20503f4a988c40e0e21117f0f6 : SUCCESS in 2m 44s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/07c6d7c89f764a1d8f40ccdea0736cdf : SUCCESS in 48m 21s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/dc95af3708b44034b75f8fbea97b06bb : SUCCESS in 26m 49s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/9f346a759df6423a99890ff7ce48167f : SUCCESS in 28m 23s\n- zun-tempest-py3-docker-sql-2024-2 https://zuul.opendev.org/t/openstack/build/86894c5537174f33bde3b978f030f531 : SUCCESS in 40m 18s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/581df6ec86f94cebb1f39c8243a03cf8 : SUCCESS in 43m 55s","accounts_in_message":[],"_revision_number":5},{"id":"6879d6e0cfd3a549fe234e501204970b880a11ee","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-12-06 02:57:39.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2c008ed43218cd38c3eae0e1881862c0bc1e017a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 02:58:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3f5852622a1475d173642dc992bf66e1a93b246e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 04:35:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db26155a90394ff59d48c5a7e700d411\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a499d19894aa4d2fa2fd2610ab05a315 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ab5ed5664b14d44812319cfb15edfb3 : SUCCESS in 4m 55s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/1cef5ea4b2194430b90e76d58ad0eee8 : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":5},{"id":"3080f3086a4869e964e79208b06ca76bd3d09d2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 04:35:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0fb90696063d7141f8bdfcb7cb0f24476bef220e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 04:37:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ad54762f1bc49c18e4b143974bc269e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8af82174854c5dac41fd6795e77759 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"069b9efbebef2df2c104e9acf3ab98518cf1595d","revisions":{"d100eaf9c11618b319146cd917441994419dfa51":{"kind":"REWORK","_number":1,"created":"2024-04-26 00:13:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/917074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/74/917074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/1"}}},"commit":{"parents":[{"commit":"8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1","subject":"Add stable/2023.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-18 23:33:53.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-26 00:13:21.000000000","tz":-420},"subject":"Add stable/2023.2|2024.1 jobs on master gate","message":"Add stable/2023.2|2024.1 jobs on master gate\n\nAs 2023.2 and 2024.1 is released, we should add its job on master\ngate to keep branchless tempest plugins compatible to stable branch.\n\nRemoving Yoga job as that is in unmaintained phase.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I9b2e2ba2336d285cf6ac35fb1ebeb78cf576a872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d100eaf9c11618b319146cd917441994419dfa51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d100eaf9c11618b319146cd917441994419dfa51"}]},"branch":"refs/heads/master"},"b959d771475c447180a827b1dd6a2656a3093140":{"kind":"REWORK","_number":2,"created":"2024-09-06 17:50:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/917074/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/74/917074/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/2"}}},"commit":{"parents":[{"commit":"8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1","subject":"Add stable/2023.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-18 23:33:53.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-09-06 17:50:23.000000000","tz":0},"subject":"Add stable/2023.2|2024.1 jobs on master gate","message":"Add stable/2023.2|2024.1 jobs on master gate\n\nAs 2023.2 and 2024.1 is released, we should add its job on master\ngate to keep branchless tempest plugins compatible to stable branch.\n\nRemoving Yoga, zed job as that is in unmaintained phase.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I9b2e2ba2336d285cf6ac35fb1ebeb78cf576a872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b959d771475c447180a827b1dd6a2656a3093140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b959d771475c447180a827b1dd6a2656a3093140"}]},"branch":"refs/heads/master"},"170b191e8235499b5d55c375bd88e20b678ebc09":{"kind":"REWORK","_number":3,"created":"2024-10-24 18:37:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/917074/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/74/917074/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/3"}}},"commit":{"parents":[{"commit":"8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1","subject":"Add stable/2023.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-18 23:33:53.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-24 18:37:49.000000000","tz":-420},"subject":"Update stable branch jobs on master gate","message":"Update stable branch jobs on master gate\n\nAs 2023.2, 2024.1, and 2024.2 is released, we should add\nits job on master gate to keep branchless tempest plugins\ncompatible to stable branch.\n\nRemoving Yoga, zed job as that is in unmaintained phase.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I9b2e2ba2336d285cf6ac35fb1ebeb78cf576a872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/170b191e8235499b5d55c375bd88e20b678ebc09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/170b191e8235499b5d55c375bd88e20b678ebc09"}]},"branch":"refs/heads/master"},"93c6e2210ca4046e7656e444e11b5bc035a9a2ee":{"kind":"REWORK","_number":4,"created":"2024-11-06 19:52:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/917074/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/74/917074/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/4"}}},"commit":{"parents":[{"commit":"8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1","subject":"Add stable/2023.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-18 23:33:53.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-06 19:52:45.000000000","tz":0},"subject":"Update stable branch jobs on master gate","message":"Update stable branch jobs on master gate\n\nAs 2023.2, 2024.1, and 2024.2 is released, we should add\nits job on master gate to keep branchless tempest plugins\ncompatible to stable branch.\n\nRemoving Yoga, zed, and 2023.1 job as that is in unmaintained phase.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I9b2e2ba2336d285cf6ac35fb1ebeb78cf576a872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/93c6e2210ca4046e7656e444e11b5bc035a9a2ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/93c6e2210ca4046e7656e444e11b5bc035a9a2ee"}]},"branch":"refs/heads/master"},"069b9efbebef2df2c104e9acf3ab98518cf1595d":{"kind":"REWORK","_number":5,"created":"2024-12-05 11:34:00.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/74/917074/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/74/917074/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/74/917074/5"}}},"commit":{"parents":[{"commit":"8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1","subject":"Add stable/2023.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-18 23:33:53.000000000","tz":-420},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2024-12-05 11:33:00.000000000","tz":0},"subject":"Update stable branch jobs on master gate","message":"Update stable branch jobs on master gate\n\nAs 2023.2, 2024.1, and 2024.2 is released, we should add\nits job on master gate to keep branchless tempest plugins\ncompatible to stable branch.\n\nRemoving Yoga, zed, and 2023.1 job as that is in unmaintained phase.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nIn addition, use \u0027block-storage\u0027 service type to detect cinder.\n\nChange-Id: I9b2e2ba2336d285cf6ac35fb1ebeb78cf576a872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/069b9efbebef2df2c104e9acf3ab98518cf1595d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/069b9efbebef2df2c104e9acf3ab98518cf1595d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
