)]}'
{"id":"openstack%2Fzun-tempest-plugin~937103","triplet_id":"openstack%2Fzun-tempest-plugin~master~I0ab49c8952ff42dd520e15550ac2a333311e825b","project":"openstack/zun-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2025-03-18 09:18:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0ab49c8952ff42dd520e15550ac2a333311e825b","subject":"Use \u0027block-storage\u0027 service type to detect cinder","status":"ABANDONED","created":"2024-12-05 10:24:49.000000000","updated":"2025-03-18 09:18:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bab00f2c8c867e5afd4523cf7514e61f9d24145b","_number":937103,"virtual_id_number":937103,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-05 11:15:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-05 11:15:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"213bc86a644c467f65b7b52cbf242183a593a829","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-12-05 10:24:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31dcc24f1d1ce86e90fb5c3b284bde7c09c94d0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 11:15:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6e5ea1317934041b70853094f9427d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a23c4a017b0c4b58a13211c9b4baee5c : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9fe32e64697407987361906ccb8a586 : SUCCESS in 3m 10s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2ccdcf8f96024f0681dbf60bb0b3f4b3 : SUCCESS in 37m 18s\n- zun-tempest-py3-docker-sql-yoga https://zuul.opendev.org/t/openstack/build/48e8f6f63b044820a4253d6d3defde61 : FAILURE in 5m 19s\n- zun-tempest-py3-docker-sql-zed https://zuul.opendev.org/t/openstack/build/98bce7ef9c7343c9b1c06c33b4952ee7 : FAILURE in 6m 31s\n- zun-tempest-py3-docker-sql-2023-1 https://zuul.opendev.org/t/openstack/build/55a571f4c81c479aa272449af16c4f1a : FAILURE in 27m 51s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/89b90dcbd1444ac8a9e271ccf3a09f8e : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":1},{"id":"bab00f2c8c867e5afd4523cf7514e61f9d24145b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-03-18 09:18:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b13fe44c2db50d5fe229b76b6de07f1e5a51fe3","revisions":{"9b13fe44c2db50d5fe229b76b6de07f1e5a51fe3":{"kind":"REWORK","_number":1,"created":"2024-12-05 10:24:49.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/03/937103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/03/937103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/03/937103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/03/937103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/03/937103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/03/937103/1"}}},"commit":{"parents":[{"commit":"8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1","subject":"Add stable/2023.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8a0b3cf100c00c5ddbe13bc97b42f16a2bff95c1"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2024-12-05 10:23:59.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2024-12-05 10:24:47.000000000","tz":0},"subject":"Use \u0027block-storage\u0027 service type to detect cinder","message":"Use \u0027block-storage\u0027 service type to detect cinder\n\nChange-Id: I0ab49c8952ff42dd520e15550ac2a333311e825b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9b13fe44c2db50d5fe229b76b6de07f1e5a51fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9b13fe44c2db50d5fe229b76b6de07f1e5a51fe3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
