)]}'
{"id":"openstack%2Fzun-tempest-plugin~948818","triplet_id":"openstack%2Fzun-tempest-plugin~master~Id60fde3eabadd8a3d25715f5ca310be6881b3872","project":"openstack/zun-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-29 19:37:16.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-29 20:35:57.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-29 20:35:39.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Id60fde3eabadd8a3d25715f5ca310be6881b3872","subject":"Ensure that images for tempest tests configurable","status":"MERGED","created":"2025-05-05 17:28:29.000000000","updated":"2026-08-29 21:16:23.000000000","submitted":"2026-08-29 21:15:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948818","meta_rev_id":"ada0fca07c37f879b060a1c1813fee35d5325011","_number":948818,"virtual_id_number":948818,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-29 21:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2026-08-29 19:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-29 20:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-29 20:35:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-05 18:01:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-28 02:41:56.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"},{"updated":"2026-06-26 23:04:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-06-26 23:04:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2026-06-26 23:04:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"3e0bad1b5485c2c6724dbb8ee0b707f90d8596c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-05 17:28:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d26471cd1f8f4035fee82e18ba044f700975b88e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 18:01:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ff0e1e3c77a4ff692134ffff521ffd3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8323252e85b640f48c5fe8eb15b92155 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a02d356c9c4e66a2a9510d9bc44293 : FAILURE in 3m 03s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/69a0ac35341449f4adedea649bab79be : FAILURE in 19m 48s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/3244ad3cfb0d4c1cb267c27871e77259 : FAILURE in 25m 02s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/38e50c44ad7c46b6b03c2f1d52be02b9 : FAILURE in 26m 17s\n- zun-tempest-py3-docker-sql-2024-2 https://zuul.opendev.org/t/openstack/build/d8f0f2bd6b444559a5cf318fc7528b90 : FAILURE in 18m 51s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f5e4ad51888e4ebabc9b26e756ca29c9 : FAILURE in 28m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e9a704a770e0820c91966905f7bd76428651bc3d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-06-28 02:41:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f385c644520dd4704a7832c29614e26713cf0e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-28 03:20:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a1dd4a65f3f43fea4d659f394d02930\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2086f1cf3a594feb98595d0e562f9d03 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6978b3bb1d3845bcb73858649cd24141 : FAILURE in 4m 11s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/24b39627d04f41b083b0d8a136917f73 : FAILURE in 29m 38s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/708ef4f0f0e041c5a680dde32c2584df : FAILURE in 28m 27s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/8c286472ad9242db9f26add729fa393a : FAILURE in 25m 02s\n- zun-tempest-py3-docker-sql-2024-2 https://zuul.opendev.org/t/openstack/build/f0052b885e144071ae850df9f91e8066 : FAILURE in 25m 46s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/51b445d6518c4014bff4085c143e2a63 : FAILURE in 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"30bd25b06a5f5c99250bb85642eb61143e2905df","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-06-30 06:08:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6deba7be4c5e7a4a4b2eefa4a790ea52164c0db9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 06:45:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32d7b974df644407b9851ae52fdedc0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04cb1d44d9ec49b19489581ed08fb822 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b3c6f44b644c24b0f553d7d4a8c8ef : FAILURE in 3m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/789910675de6455e94c757f6d33fa9db : FAILURE in 27m 38s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/c6b1238420c6485bb2900fd3bf441f51 : FAILURE in 26m 40s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/960a7fb412a84b53b2b43f72c5f20eb3 : FAILURE in 25m 11s\n- zun-tempest-py3-docker-sql-2024-2 https://zuul.opendev.org/t/openstack/build/b17d45bd8c5946bc95378080fa86e34e : FAILURE in 25m 10s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/845fcefb482f441b95244a1fa4fb8536 : FAILURE in 32m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4d8a6027f0de23b4af3d7008f32c8e2bd2cf5a56","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-07-01 02:21:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3f3f1c627dd3bb31c846efc2d995fed728ac247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 03:17:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/381475263d074173962bbe7b16d1c504\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a757e31fbe40425c89271a3dd6d350d2 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405c987e3c1346679a4a00b4cc744a74 : FAILURE in 2m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/be07ecb8c2264c3dbdbf8f6c8385640e : FAILURE in 21m 51s\n- zun-tempest-py3-docker-sql-2023-2 https://zuul.opendev.org/t/openstack/build/a2fd281ea4284baf8cea1c275c4e666c : FAILURE in 29m 15s\n- zun-tempest-py3-docker-sql-2024-1 https://zuul.opendev.org/t/openstack/build/6190f62ded96479fbf220e7afd235135 : FAILURE in 27m 45s\n- zun-tempest-py3-docker-sql-2024-2 https://zuul.opendev.org/t/openstack/build/8f3544739b604f43a846c7357ffd1cba : FAILURE in 25m 48s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2f2aeb97c07c4f2eafdfb4ae08ed5c56 : FAILURE in 33m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5a8639b1fd5689c6696b76c9f6d676c98e1184a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 14:22:12.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3bb50dd31dabca829eb42aa8697703e1f380528d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 14:22:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"43d4e357f96a40eebd51226ba67c00e66108ccfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 14:23:58.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5cb38a101da016e410872d899ed5511ccc64840d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 15:04:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/91dc20568a9c404bbb789b2384b7bd2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38edc98da137476992994da82628fb8f : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c494c97147441e7a73604b880db2ca8 : FAILURE in 4m 24s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/47e2084c55c8406a880e352f43d754bd : FAILURE in 26m 34s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/c9999c2c35e44bf8819c72faaa501293 : FAILURE in 31m 18s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/95fc376b706c41f3bc8efc70e758f81e : FAILURE in 15m 12s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/54de2b00d08c4b6e8d9a38fcff62cd2d : FAILURE in 32m 21s","accounts_in_message":[],"_revision_number":4},{"id":"cb171cf5c047a3d5679e6ea1eb1fe923c2e43eb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 16:28:26.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ad2fcc2dec5770c220f877ef50d3128ceedc6cae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 17:17:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/115d7f2bb072417fb787a19731028eea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5373e554a4df43d89f6023b5dc6f0ab8 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7e815ebd3e40e299c5ef777e4609d4 : FAILURE in 4m 55s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b230bcd4425b4f52a8103faf468f8dac : FAILURE in 15m 37s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/be7023cd53ae448ca4b9bde143e80353 : FAILURE in 26m 23s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/e5cba75ca3374b7ebfdce86615c58db5 : FAILURE in 31m 58s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/3b43bf15c7d2477d91a2c54e1707cba8 : FAILURE in 32m 59s","accounts_in_message":[],"_revision_number":5},{"id":"d93402318c343c08baca6216039a02ccf8994f78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 20:36:21.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"837c6b30b0ab9f353790f0f5e497b8d5b0906a6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 21:05:45.000000000","message":"Patch Set 6: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1070092b898843e6a0c0d4e46d4b22af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19968b461d84d939f64f2ac14445aee : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/864d9aa19c1d438ba6c6e29a32b11895 : FAILURE in 2m 49s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a0acd59e387f405abc783131c3dd6b76 : FAILURE in 25m 45s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/e53bfd4c419a4591a961ab13bdaf7f59 : FAILURE in 25m 29s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/5c921295cb5f47a68b431167beac363a : FAILURE in 15m 03s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/3f126af502ff44899068cc8d4b12b22a : FAILURE in 14m 40s","accounts_in_message":[],"_revision_number":6},{"id":"a9b274f6e61d704c52453761aa73d9f7f8690952","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 21:06:33.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"05f61ec71c95ced45b06f167a2781c155667ac3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 21:09:11.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"2f159c59e6261d109848f6e32cb43920bdbddb2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 21:13:23.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"7ca74ccc934b2703cdd1b97a77beff4b213b8f7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 21:16:46.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"1793c7fd6fea93f86dc5d071e3b044aaa6c5526a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 21:56:27.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"acb2cee8b81a9f6b24cb761dc4fe45d5365c9f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 22:00:01.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"4e3abeaccd7ac1bd646a0fd3c0860378ada51903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 22:01:15.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"37a20dbc52ae4b2120a83b57bf670d36c7f974bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 22:55:53.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/220fc7b8d535447c9b45b68853e21dd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff4be8818037477e859ebb7066c5a415 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/803371ae54004c0d99c705c285c2d436 : SUCCESS in 2m 42s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b8f5a11a0a2546ec8f387b73a837c413 : SUCCESS in 44m 59s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/97488ce2947840d78247cbeffa645967 : SUCCESS in 48m 17s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/d63ebdf0309a47e1b8983f7df1db0e9a : SUCCESS in 43m 09s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f5813c7589b24ecfa3f1509dfbb381df : SUCCESS in 46m 06s","accounts_in_message":[],"_revision_number":13},{"id":"6da7136f3a83d88a935d16dcd95ce8fa60a1810f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 23:04:28.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"175df2ed22c09f5d63caa1e07eda8bd2b9d29eb5","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2026-07-27 07:20:28.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"f2a6fdbf9a482f0c93965b3f0db5eeb298092339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-17 11:08:14.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"b4d0ef9bcc6e9959a49cc286ca8452a64eacdee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 11:45:12.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/061502dadde847f1867178295eb2cda9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c1f19c6e19e4897957aa99ad5e21d98 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da21787856904ea3bd16318fa321a889 : SUCCESS in 3m 15s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/092862a4bb1c421e8486c882b503ee78 : FAILURE in 34m 47s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/44210db478184a4fbc4dacf0784d6c09 : SUCCESS in 30m 18s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/48ae2d5677bd4365aa8fd54f838e7220 : SUCCESS in 35m 21s\n- zun-tempest-py3-docker-sql-2026-1 https://zuul.opendev.org/t/openstack/build/f90d7d527553417b820ffb49acb13c9c : SUCCESS in 25m 09s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f8c32f79201c46e1876e247b0eac12f4 : SUCCESS in 29m 38s","accounts_in_message":[],"_revision_number":14},{"id":"5b55ec03f282fd4b816ca8dedaa900e56ebb4c35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 19:37:07.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"b6a58da28fcae6ee9063d4f064ea43c94428bb0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 20:30:19.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eac526d6509f493a939a602a6dfc0a64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cff7c0c50cb49a191278cc1a6af707d : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f78ac4acf042a0be2798f05e39d9ed : SUCCESS in 2m 57s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/be8e9a2686ba410782f2aec96472514f : SUCCESS in 43m 00s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/1fa07e84ab3e43f2901186a4eb256ce8 : SUCCESS in 31m 26s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/30bf709a5d4c420daaaded8ebf76aaef : SUCCESS in 42m 39s\n- zun-tempest-py3-docker-sql-2026-1 https://zuul.opendev.org/t/openstack/build/5f4dcef525be431b8fce554cf921887e : SUCCESS in 46m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/0bc249dd963542e39fbba5fd573f6dfa : SUCCESS in 47m 29s","accounts_in_message":[],"_revision_number":15},{"id":"5ab63aa87c4ddc3381cd3e1c34e9a44cd41fc693","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 20:35:39.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"f281023584f6d3d69ad1565982bcf138991d654f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 20:35:54.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ce58279928ef6a56046b89f01b96bde530679ed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 20:36:04.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"7965c086086e5ae99053d6466a46cd4be6d98bd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 21:15:15.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/371bea2617214919b5de57962be333f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c744bcc4cc824666b036ebac6203cd55 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3350c44b9a844317bb2ad59e16549675 : SUCCESS in 3m 04s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/fd487a26a3804e509e15af3372065f0f : SUCCESS in 24m 37s","accounts_in_message":[],"_revision_number":15},{"id":"452ba21b0a8faab38a8863ca8b5ea75baed9ef84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 21:15:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"ada0fca07c37f879b060a1c1813fee35d5325011","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 21:16:23.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd8f3476d7714e0fab243e9dc4b9e3dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80584f3270c440d18cc234050efe57b1 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"1c8845645f750337a7b9bf13a66af52562ada80a","revisions":{"a008bd56b7ca8150bad5afebc479b6a5a3a112ca":{"kind":"REWORK","_number":1,"created":"2025-05-05 17:28:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/948818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/1"}}},"commit":{"parents":[{"commit":"069b9efbebef2df2c104e9acf3ab98518cf1595d","subject":"Update stable branch jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/069b9efbebef2df2c104e9acf3ab98518cf1595d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:27.000000000","tz":120},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a008bd56b7ca8150bad5afebc479b6a5a3a112ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a008bd56b7ca8150bad5afebc479b6a5a3a112ca"}]},"branch":"refs/heads/master"},"8e2d98fcc0aec14628396c824f6140c8338dadc9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-26 14:22:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/2"}}},"commit":{"parents":[{"commit":"069b9efbebef2df2c104e9acf3ab98518cf1595d","subject":"Update stable branch jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/069b9efbebef2df2c104e9acf3ab98518cf1595d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 14:22:12.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8e2d98fcc0aec14628396c824f6140c8338dadc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8e2d98fcc0aec14628396c824f6140c8338dadc9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b05266d09415940fe1d4ade489bcf5789b44eead":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-06-26 14:22:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/3"}}},"commit":{"parents":[{"commit":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","subject":"Remove EOl stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/60235d87783de40d2b80bbb2df3e2ea22e3b0fed"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 14:22:39.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b05266d09415940fe1d4ade489bcf5789b44eead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b05266d09415940fe1d4ade489bcf5789b44eead"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"069b9efbebef2df2c104e9acf3ab98518cf1595d","ours":"8e2d98fcc0aec14628396c824f6140c8338dadc9","theirs":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","merge_strategy":"recursive","contains_conflicts":false}},"471a40c92285e7e1800b86653ebdd0c0c7091fe8":{"kind":"REWORK","_number":4,"created":"2026-06-26 14:23:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/4"}}},"commit":{"parents":[{"commit":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","subject":"Remove EOl stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/60235d87783de40d2b80bbb2df3e2ea22e3b0fed"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 14:23:58.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/471a40c92285e7e1800b86653ebdd0c0c7091fe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/471a40c92285e7e1800b86653ebdd0c0c7091fe8"}]},"branch":"refs/heads/master"},"280b7d6bcd92555a0f123a8fa60b457bcb232ed6":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-06-26 16:28:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/5"}}},"commit":{"parents":[{"commit":"59699b11e8096304a8f095e84fa0b5b8da29834e","subject":"Bump docker api version to 1.40","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/59699b11e8096304a8f095e84fa0b5b8da29834e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 16:28:26.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/280b7d6bcd92555a0f123a8fa60b457bcb232ed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/280b7d6bcd92555a0f123a8fa60b457bcb232ed6"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","ours":"471a40c92285e7e1800b86653ebdd0c0c7091fe8","theirs":"59699b11e8096304a8f095e84fa0b5b8da29834e","merge_strategy":"recursive","contains_conflicts":false}},"149e76512761a79ff697100403754c0e82255f71":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-06-26 20:36:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/6"}}},"commit":{"parents":[{"commit":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a1e8e03c74a8ababfc61c419176fd06dfd7103ac"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 20:36:21.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/149e76512761a79ff697100403754c0e82255f71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/149e76512761a79ff697100403754c0e82255f71"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"59699b11e8096304a8f095e84fa0b5b8da29834e","ours":"280b7d6bcd92555a0f123a8fa60b457bcb232ed6","theirs":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","merge_strategy":"recursive","contains_conflicts":false}},"f869d9a7ad8d20be3abf3ee20945013f8c5d90c4":{"kind":"REWORK","_number":7,"created":"2026-06-26 21:06:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/7"}}},"commit":{"parents":[{"commit":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a1e8e03c74a8ababfc61c419176fd06dfd7103ac"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 21:06:33.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f869d9a7ad8d20be3abf3ee20945013f8c5d90c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f869d9a7ad8d20be3abf3ee20945013f8c5d90c4"}]},"branch":"refs/heads/master"},"6c0672473e4995e8a9751a56a07523a4cb669a81":{"kind":"REWORK","_number":8,"created":"2026-06-26 21:09:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/8"}}},"commit":{"parents":[{"commit":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a1e8e03c74a8ababfc61c419176fd06dfd7103ac"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 21:09:11.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6c0672473e4995e8a9751a56a07523a4cb669a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6c0672473e4995e8a9751a56a07523a4cb669a81"}]},"branch":"refs/heads/master"},"4fd214ff676995954d7dd6cfb00c4cbe4122e91a":{"kind":"REWORK","_number":9,"created":"2026-06-26 21:13:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/9"}}},"commit":{"parents":[{"commit":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a1e8e03c74a8ababfc61c419176fd06dfd7103ac"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 21:13:23.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/4fd214ff676995954d7dd6cfb00c4cbe4122e91a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/4fd214ff676995954d7dd6cfb00c4cbe4122e91a"}]},"branch":"refs/heads/master"},"71810d6073b85e38560569f686b30e6f7d41dd31":{"kind":"REWORK","_number":10,"created":"2026-06-26 21:16:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/10"}}},"commit":{"parents":[{"commit":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a1e8e03c74a8ababfc61c419176fd06dfd7103ac"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 21:16:45.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/71810d6073b85e38560569f686b30e6f7d41dd31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/71810d6073b85e38560569f686b30e6f7d41dd31"}]},"branch":"refs/heads/master"},"8454e9f3d2cc05d43640a0affe1028da056d650b":{"kind":"REWORK","_number":11,"created":"2026-06-26 21:56:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/11"}}},"commit":{"parents":[{"commit":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a1e8e03c74a8ababfc61c419176fd06dfd7103ac"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 21:56:27.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8454e9f3d2cc05d43640a0affe1028da056d650b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8454e9f3d2cc05d43640a0affe1028da056d650b"}]},"branch":"refs/heads/master"},"1e129f9e46df46f01ec932e722928120a40c96b4":{"kind":"REWORK","_number":12,"created":"2026-06-26 22:00:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/12"}}},"commit":{"parents":[{"commit":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a1e8e03c74a8ababfc61c419176fd06dfd7103ac"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 22:00:01.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1e129f9e46df46f01ec932e722928120a40c96b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1e129f9e46df46f01ec932e722928120a40c96b4"}]},"branch":"refs/heads/master"},"85ba05746bc01f804865dd5866d36a55198a1a6f":{"kind":"REWORK","_number":13,"created":"2026-06-26 22:01:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/13"}}},"commit":{"parents":[{"commit":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a1e8e03c74a8ababfc61c419176fd06dfd7103ac"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 22:01:15.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/85ba05746bc01f804865dd5866d36a55198a1a6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/85ba05746bc01f804865dd5866d36a55198a1a6f"}]},"branch":"refs/heads/master"},"4c955d3a0bab4343c0fb370a59061ea93fa7f8ff":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-08-17 11:08:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/14"}}},"commit":{"parents":[{"commit":"69b7e4f7ecd106f97b784cd675f633b4b2b90e88","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/69b7e4f7ecd106f97b784cd675f633b4b2b90e88"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-17 11:08:14.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/4c955d3a0bab4343c0fb370a59061ea93fa7f8ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/4c955d3a0bab4343c0fb370a59061ea93fa7f8ff"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a1e8e03c74a8ababfc61c419176fd06dfd7103ac","ours":"85ba05746bc01f804865dd5866d36a55198a1a6f","theirs":"69b7e4f7ecd106f97b784cd675f633b4b2b90e88","merge_strategy":"recursive","contains_conflicts":false}},"1c8845645f750337a7b9bf13a66af52562ada80a":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-08-29 19:37:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/948818/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/18/948818/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/18/948818/15"}}},"commit":{"parents":[{"commit":"1041a6b7b8230a36c37cb91f37177aed0ca85880","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1041a6b7b8230a36c37cb91f37177aed0ca85880"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 17:28:08.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 19:37:07.000000000","tz":0},"subject":"Ensure that images for tempest tests configurable","message":"Ensure that images for tempest tests configurable\n\nWith severe rate limits applied to images pull on Dockerhub,\nthere should be an option to use a custom path to images for\nTempest tests. This should allow configuration and usage of 3rd\nparty registries, like quay.io or others.\n\nChange-Id: Id60fde3eabadd8a3d25715f5ca310be6881b3872\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"69b7e4f7ecd106f97b784cd675f633b4b2b90e88","ours":"4c955d3a0bab4343c0fb370a59061ea93fa7f8ff","theirs":"1041a6b7b8230a36c37cb91f37177aed0ca85880","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
