)]}'
{"id":"openstack%2Fzun-tempest-plugin~991195","triplet_id":"openstack%2Fzun-tempest-plugin~master~Id7b0bc4982b9a69a7516088ea66a879554370fa6","project":"openstack/zun-tempest-plugin","branch":"master","topic":"drop-py3-10","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-07-27 07:24:19.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-08-16 09:18:49.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-16 09:18:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7b0bc4982b9a69a7516088ea66a879554370fa6","subject":"Drop python 3.10","status":"MERGED","created":"2026-06-02 19:20:45.000000000","updated":"2026-08-16 09:20:27.000000000","submitted":"2026-08-16 09:18:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991195-drop-py3-10","meta_rev_id":"dad90ed5c6ab84b6317d054a16e5bcd1ca4bb983","_number":991195,"virtual_id_number":991195,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-16 09:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2026-07-27 07:24:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-26 15:20:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2026-07-27 07:24:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-02 20:17:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 14:21:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-06-26 14:21:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"ac30efc4a2d0a6db46a22f81ecc884c181c9fb7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-02 19:20:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abd96705cb79756da8709b3c6a60a90686db9260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 20:17:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54cd0961dc2846ba89cd5fc828a37a5a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/edc73d11a33d4f58a3a10fa20a344b89 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aeb8a4ea8b84f63a05e3481352791d6 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac8bfc065ee446bbe4990d1bd56a7c2 : SUCCESS in 4m 14s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/040f3a261de949ae83a83235a92ef98b : SUCCESS in 40m 15s\n- zun-tempest-py3-docker-sql-2024-2 https://zuul.opendev.org/t/openstack/build/b6479db703534b9a8ae8db098f82720e : FAILURE in 8m 57s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/d45a2abe24e247ae9b1fca1de4be2db6 : FAILURE in 27m 04s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/e520bc4ca8bf4b87a64e1d14c59d205f : SUCCESS in 46m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/6ec2a918629b42018fbb48300fcaad21 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5ae2539186e587b040e9eacad5b9cc5bef00ac2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 14:16:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":2},{"id":"57d461670b2dd9346b3d7ecda9784a896759961b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 14:21:06.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b6ab3ddea6fee432ba1c1d72a01a4b3f3b8fdb1a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 14:21:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b54c6dcf63d180494bba330bd16e3308e12a956b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 15:15:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c055ac701ae4d8983fb376f5799549b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/06c238ef58cd4190b64554f6fccb7b9b : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6081152e45cc41c98414c2f11db73c81 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71f8d611da824c9aaf337e20af1140fd : SUCCESS in 2m 09s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ad238dc157114c678af7b0a91ee0559d : FAILURE in 26m 53s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/43c63a58b7504dbb87098fa750db532e : SUCCESS in 44m 30s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/8920c3f438284bd99052a9883a3e7f9a : SUCCESS in 45m 28s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d9f63c33299241bca666231639596bd8 : SUCCESS in 47m 21s","accounts_in_message":[],"_revision_number":3},{"id":"928dbca8c31d8bfa1cb540ad5e2dd50cb7cfe714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 15:20:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"592f0dbc779a1baa3367d616d35d253ee81c33c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 16:21:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a8fe8d81c924d6c98e8ca060d737ee1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1c035ac155a04391890f82466777fdef : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783e7d8e1b3f4b83a3b51b81eaba46f9 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8ffe51772c4b5195770b111eb9f0ce : SUCCESS in 3m 24s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2790fa7fe34d41d387733c65db03a967 : SUCCESS in 48m 27s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/2a694760d9e54aae88c700f95bd883c8 : SUCCESS in 46m 52s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/391961ff10674127b8be08a16899bd82 : SUCCESS in 44m 58s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2d42bb944aab430383ed0d0ddcef84da : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":4},{"id":"1da862a01d283441044f04a44980843cf7c3a90b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2026-07-27 07:24:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"85bc0ace235f4a3beb0aac321eacfe6c578efaae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 08:03:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0bcd2c3d222f2cdd3e4bc6798e1b77437eaadf07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 09:18:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/606dbb0046b045f59f8aac5fa6e8f377\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/63afb8dc63f244d3ac025863b98bd0f0 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e12bae463f044028d0e157db497790c : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1397d7e4ad849fd95f485ebfd6f9851 : SUCCESS in 2m 53s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/5b6a7615c15a4337989bff63b4ab2a20 : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":4},{"id":"253a408968748ca28c4f0775c883063ebad15c9d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 09:18:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"dad90ed5c6ab84b6317d054a16e5bcd1ca4bb983","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 09:20:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c724039941c4b72aa3b0ea4777a0901\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4602ab2f87094bbc9a09d0563fc13493 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d42e8f8a03afacca8e660a89b739df20c2e44184","revisions":{"994fd8e47294577adce4af29def5a58564c2e6bc":{"kind":"REWORK","_number":1,"created":"2026-06-02 19:20:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/991195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/95/991195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/1"}}},"commit":{"parents":[{"commit":"d8595f07c74f1314f15beb8e42ca68d2b552a036","subject":"Update stable branch jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d8595f07c74f1314f15beb8e42ca68d2b552a036"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-02 18:33:23.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-02 18:33:23.000000000","tz":0},"subject":"Drop python 3.10","message":"Drop python 3.10\n\nDue to constraints update(rpds-py) in requirements, Tempest master stopped\nworking on python 3.10.\n\n- https://review.opendev.org/c/openstack/requirements/+/990790\n\nDue to same reason, Tempest needs to drop the python 3.10 support[1].\nBecause Tempest master cannot support py3.10 anymore, it needs to be\ndropped from tempest plugins also.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/991161\n\nChange-Id: Id7b0bc4982b9a69a7516088ea66a879554370fa6\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/994fd8e47294577adce4af29def5a58564c2e6bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/994fd8e47294577adce4af29def5a58564c2e6bc"}]},"branch":"refs/heads/master"},"feb74ef5c8fca3496d441652349fb37421bad643":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-26 14:16:58.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/991195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/95/991195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/2"}}},"commit":{"parents":[{"commit":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","subject":"Remove EOl stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/60235d87783de40d2b80bbb2df3e2ea22e3b0fed"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-02 18:33:23.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-26 14:16:58.000000000","tz":0},"subject":"Drop python 3.10","message":"Drop python 3.10\n\nDue to constraints update(rpds-py) in requirements, Tempest master stopped\nworking on python 3.10.\n\n- https://review.opendev.org/c/openstack/requirements/+/990790\n\nDue to same reason, Tempest needs to drop the python 3.10 support[1].\nBecause Tempest master cannot support py3.10 anymore, it needs to be\ndropped from tempest plugins also.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/991161\n\nChange-Id: Id7b0bc4982b9a69a7516088ea66a879554370fa6\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/feb74ef5c8fca3496d441652349fb37421bad643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/feb74ef5c8fca3496d441652349fb37421bad643"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d8595f07c74f1314f15beb8e42ca68d2b552a036","ours":"994fd8e47294577adce4af29def5a58564c2e6bc","theirs":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","merge_strategy":"recursive","contains_conflicts":false}},"3a69347d159bda5fb762bdd2864aad47a849617b":{"kind":"REWORK","_number":3,"created":"2026-06-26 14:21:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/991195/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/95/991195/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/3"}}},"commit":{"parents":[{"commit":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","subject":"Remove EOl stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/60235d87783de40d2b80bbb2df3e2ea22e3b0fed"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-02 18:33:23.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 14:21:06.000000000","tz":0},"subject":"Drop python 3.10","message":"Drop python 3.10\n\nDue to constraints update(rpds-py) in requirements, Tempest master stopped\nworking on python 3.10.\n\n- https://review.opendev.org/c/openstack/requirements/+/990790\n\nDue to same reason, Tempest needs to drop the python 3.10 support[1].\nBecause Tempest master cannot support py3.10 anymore, it needs to be\ndropped from tempest plugins also.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/991161\n\nChange-Id: Id7b0bc4982b9a69a7516088ea66a879554370fa6\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/3a69347d159bda5fb762bdd2864aad47a849617b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/3a69347d159bda5fb762bdd2864aad47a849617b"}]},"branch":"refs/heads/master"},"d42e8f8a03afacca8e660a89b739df20c2e44184":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-26 15:20:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/991195/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/95/991195/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/95/991195/4"}}},"commit":{"parents":[{"commit":"59699b11e8096304a8f095e84fa0b5b8da29834e","subject":"Bump docker api version to 1.40","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/59699b11e8096304a8f095e84fa0b5b8da29834e"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-02 18:33:23.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 15:20:10.000000000","tz":0},"subject":"Drop python 3.10","message":"Drop python 3.10\n\nDue to constraints update(rpds-py) in requirements, Tempest master stopped\nworking on python 3.10.\n\n- https://review.opendev.org/c/openstack/requirements/+/990790\n\nDue to same reason, Tempest needs to drop the python 3.10 support[1].\nBecause Tempest master cannot support py3.10 anymore, it needs to be\ndropped from tempest plugins also.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/991161\n\nChange-Id: Id7b0bc4982b9a69a7516088ea66a879554370fa6\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d42e8f8a03afacca8e660a89b739df20c2e44184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d42e8f8a03afacca8e660a89b739df20c2e44184"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","ours":"3a69347d159bda5fb762bdd2864aad47a849617b","theirs":"59699b11e8096304a8f095e84fa0b5b8da29834e","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
