)]}'
{"id":"openstack%2Fzun-tempest-plugin~995055","triplet_id":"openstack%2Fzun-tempest-plugin~master~I1707c988be3119470882d6f29e503799cd923448","project":"openstack/zun-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-16 08:02:39.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-16 09:18:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1707c988be3119470882d6f29e503799cd923448","subject":"Bump docker api version to 1.40","status":"MERGED","created":"2026-06-26 13:00:09.000000000","updated":"2026-08-16 09:19:35.000000000","submitted":"2026-08-16 09:18:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995055","meta_rev_id":"5821cacaa1040babb1ad6aad36e99b0de082e551","_number":995055,"virtual_id_number":995055,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-16 09:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2026-08-16 08:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2026-08-16 08:02:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 14:07:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 14:18:26.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"bd9f3c4f53480cfb78bd82b1417391ff1b991410","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 13:00:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cd170026ba06a7c201e44f37be13c7a7479363c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 13:02:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f23e84d8ee37c39ea2f7962700e754774bb52b02","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 13:02:18.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"36db76fbde0304632304cdd409147c9f8a34d306","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 13:02:20.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"b5e4a4b18143f5dab640085c5a17fd30311528b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 13:11:11.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a21f7d3c6b03248954e8d2442e9f7a757e4f3052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 14:07:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e4139542b67415f8938a63c62d49e4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72059e36e8e143d08c04bfd76b4135f0 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374ab4c731ea450884be9681bb6f4714 : SUCCESS in 2m 20s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/468f2cdc80d04f4490d50da33f323c91 : SUCCESS in 49m 25s\n- zun-tempest-py3-docker-sql-2024-2 https://zuul.opendev.org/t/openstack/build/470e3d7209424a8fa8e64f59fce88323 : FAILURE in 6m 46s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/314325cf40d94c3197145221372359ed : FAILURE in 30m 21s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/f51962ab55364bbebee96de2a0a53ded : SUCCESS in 33m 34s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f0255e7f4c4e40c6bdeeb5cba3cc7e05 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":4},{"id":"83dfd2efa905957a88a0e478f2912989a8d26f8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 14:16:25.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"dc8ef935f1fec87dec936700362d4db575f3e548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 15:16:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a192b8b51b5b451bb05e2e4111dc7540\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97cd08210b284de29dfd27cfe1a46d79 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7640dbafdaec47b6a206fbe6069b367b : SUCCESS in 3m 16s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/48d29a74c7e343dcb0608771f16d7e83 : SUCCESS in 44m 30s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/cd69320bffdc4c17911ace4cfbbdce66 : SUCCESS in 28m 02s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/6304136af3874018a6148b789b6d338b : SUCCESS in 48m 16s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/872fb2069b794f5c9c13ea9286b35a6e : SUCCESS in 52m 57s","accounts_in_message":[],"_revision_number":5},{"id":"ab30606b6e2f27f59105ae48732d728971df4fd9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2026-07-27 07:23:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"081309f072479e6f15a67715bba5e5ce37e87291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 08:17:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f74d496fb1d64da4acc1763663f87a1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbc3edf9de940b0b50fb6e4d40cdbcb : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7350f256def4b089ff384b1ad89f2fb : SUCCESS in 3m 09s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ffaca0c9ad3845479b5ebdf73eef09f5 : SUCCESS in 43m 02s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/270c11a1e5f644ae90c9fc0e4c2dbb19 : SUCCESS in 47m 04s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/16bdd724490b4496a19ec671e49aa92b : SUCCESS in 45m 25s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/e244ba9aca42476990428f1c23825ed0 : SUCCESS in 44m 53s","accounts_in_message":[],"_revision_number":5},{"id":"741adea0f91ee6d9f1a361f70b4f1ef2c90e9bdc","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2026-08-16 08:02:39.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"88e3937c2f797cfbd8b3b984c03e8f57a48f1a1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 08:03:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e3cfdd4c7e7ac228c0f5a0aaa185b83033f598dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 09:18:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74ca7af7be5b414abd42337bdbe4294d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b31d3993a07408995abf91288abc271 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3a0d1733ec4522bf05c420c62a0f6b : SUCCESS in 2m 54s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/d92a92bab422452ebcc8b73cb6de8e14 : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":5},{"id":"ae85a5e188d0fe1ce30960889d965c55abd0807d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 09:18:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"5821cacaa1040babb1ad6aad36e99b0de082e551","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 09:19:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b14835e84014dc397ee225ec9f82f55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20f951dd55b4b80a78fb353b4b274fe : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"59699b11e8096304a8f095e84fa0b5b8da29834e","revisions":{"c13cb68699e106fe46d68fc91d30322569da02ca":{"kind":"REWORK","_number":1,"created":"2026-06-26 13:00:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/995055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/55/995055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/1"}}},"commit":{"parents":[{"commit":"d8595f07c74f1314f15beb8e42ca68d2b552a036","subject":"Update stable branch jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d8595f07c74f1314f15beb8e42ca68d2b552a036"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 13:00:09.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 13:00:09.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I1707c988be3119470882d6f29e503799cd923448\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/c13cb68699e106fe46d68fc91d30322569da02ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/c13cb68699e106fe46d68fc91d30322569da02ca"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"8b043e65d8d5977e869a5d84149f9f5ccf0416b1":{"kind":"REWORK","_number":2,"created":"2026-06-26 13:02:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/995055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/55/995055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/2"}}},"commit":{"parents":[{"commit":"d8595f07c74f1314f15beb8e42ca68d2b552a036","subject":"Update stable branch jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d8595f07c74f1314f15beb8e42ca68d2b552a036"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 13:00:09.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 13:02:05.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I1707c988be3119470882d6f29e503799cd923448\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8b043e65d8d5977e869a5d84149f9f5ccf0416b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/8b043e65d8d5977e869a5d84149f9f5ccf0416b1"}]},"branch":"refs/heads/master"},"e0dae8c860587d7140111e2303dccc83d3d905ff":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-26 13:02:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/995055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/55/995055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/3"}}},"commit":{"parents":[{"commit":"d8595f07c74f1314f15beb8e42ca68d2b552a036","subject":"Update stable branch jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d8595f07c74f1314f15beb8e42ca68d2b552a036"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 13:00:09.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 13:02:18.000000000","tz":0},"subject":"Bump docker api version to 1.40","message":"Bump docker api version to 1.40\n\nChange-Id: I1707c988be3119470882d6f29e503799cd923448\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/e0dae8c860587d7140111e2303dccc83d3d905ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/e0dae8c860587d7140111e2303dccc83d3d905ff"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6f2f28491749c1e304a915f0adad87d45f60c3ad":{"kind":"REWORK","_number":4,"created":"2026-06-26 13:11:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/995055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/55/995055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/4"}}},"commit":{"parents":[{"commit":"d8595f07c74f1314f15beb8e42ca68d2b552a036","subject":"Update stable branch jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d8595f07c74f1314f15beb8e42ca68d2b552a036"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 13:00:09.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 13:11:11.000000000","tz":0},"subject":"Bump docker api version to 1.40","message":"Bump docker api version to 1.40\n\nChange-Id: I1707c988be3119470882d6f29e503799cd923448\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6f2f28491749c1e304a915f0adad87d45f60c3ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6f2f28491749c1e304a915f0adad87d45f60c3ad"}]},"branch":"refs/heads/master"},"59699b11e8096304a8f095e84fa0b5b8da29834e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-06-26 14:16:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/995055/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/55/995055/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/55/995055/5"}}},"commit":{"parents":[{"commit":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","subject":"Remove EOl stable/2024.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/60235d87783de40d2b80bbb2df3e2ea22e3b0fed"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 13:00:09.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-26 14:16:25.000000000","tz":0},"subject":"Bump docker api version to 1.40","message":"Bump docker api version to 1.40\n\nChange-Id: I1707c988be3119470882d6f29e503799cd923448\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/59699b11e8096304a8f095e84fa0b5b8da29834e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/59699b11e8096304a8f095e84fa0b5b8da29834e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d8595f07c74f1314f15beb8e42ca68d2b552a036","ours":"6f2f28491749c1e304a915f0adad87d45f60c3ad","theirs":"60235d87783de40d2b80bbb2df3e2ea22e3b0fed","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
