)]}'
{"id":"openstack%2Fzun-ui~654064","triplet_id":"openstack%2Fzun-ui~stable%2Fqueens~I873d370f210fcf45999c1b974d02b8a549649d24","project":"openstack/zun-ui","branch":"stable/queens","hashtags":[],"change_id":"I873d370f210fcf45999c1b974d02b8a549649d24","subject":"Switch back to nodeset ubuntu-xenial","status":"MERGED","created":"2019-04-20 21:40:42.000000000","updated":"2019-04-20 22:36:11.000000000","submitted":"2019-04-20 22:36:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654064-1555799771515-81340386","meta_rev_id":"5a187f3e3067a593fda256f843994e831e99d8b4","_number":654064,"virtual_id_number":654064,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-04-20 22:36:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-04-20 22:20:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-04-20 22:20:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-20 22:36:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5708a29dc5dbf53a76df8a48e0a9239445c713c9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-20 21:40:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"526eb80dce76200537d774c5c574857bdc852744","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 21:51:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/64/654064/1/check/openstack-tox-pep8/c397096/ : SUCCESS in 5m 07s\n- openstack-tox-py27 http://logs.openstack.org/64/654064/1/check/openstack-tox-py27/46f7fd4/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/64/654064/1/check/openstack-tox-py35/40d53fb/ : SUCCESS in 5m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/64/654064/1/check/build-openstack-sphinx-docs/fe3828e/html/ : SUCCESS in 4m 49s\n- nodejs-npm-run-lint http://logs.openstack.org/64/654064/1/check/nodejs-npm-run-lint/f9fbeda/ : SUCCESS in 5m 20s\n- nodejs-npm-run-test http://logs.openstack.org/64/654064/1/check/nodejs-npm-run-test/625a262/ : SUCCESS in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb2653cef747ae7bfaad98288d45b5d87a4766a9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-20 22:20:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c178b5474927f99f93da86bec824671193d32321","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 22:20:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4adde5ee9859bc9f5781c3ee1bdbe21e3a2b3582","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 22:36:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5a187f3e3067a593fda256f843994e831e99d8b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 22:36:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/64/654064/1/gate/openstack-tox-pep8/93c9bee/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/64/654064/1/gate/openstack-tox-py27/22a5ba2/ : SUCCESS in 6m 14s\n- openstack-tox-py35 http://logs.openstack.org/64/654064/1/gate/openstack-tox-py35/a9ac83f/ : SUCCESS in 7m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/64/654064/1/gate/build-openstack-sphinx-docs/d77086b/html/ : SUCCESS in 5m 29s\n- nodejs-npm-run-lint http://logs.openstack.org/64/654064/1/gate/nodejs-npm-run-lint/9c2a0ca/ : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82c1538bdf64fa0d41b6b3d08695724eb1d60f2a","revisions":{"82c1538bdf64fa0d41b6b3d08695724eb1d60f2a":{"kind":"REWORK","_number":1,"created":"2019-04-20 21:40:42.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/64/654064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-ui","ref":"refs/changes/64/654064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/64/654064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/64/654064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/64/654064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-ui refs/changes/64/654064/1"}}},"commit":{"parents":[{"commit":"3666c3364a9415af0cfd606917f35f8b7c979929","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/3666c3364a9415af0cfd606917f35f8b7c979929"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-20 21:40:13.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-20 21:40:42.000000000","tz":0},"subject":"Switch back to nodeset ubuntu-xenial","message":"Switch back to nodeset ubuntu-xenial\n\nThe default nodeset of nodejs jobs was updated to bionic, which\nbreak the stable branch jobs. This commit fixes it.\n\nChange-Id: I873d370f210fcf45999c1b974d02b8a549649d24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/82c1538bdf64fa0d41b6b3d08695724eb1d60f2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/82c1538bdf64fa0d41b6b3d08695724eb1d60f2a"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
