)]}'
{"id":"openstack%2Fzun-ui~689639","triplet_id":"openstack%2Fzun-ui~stable%2Ftrain~Ic3dd0f949c3937f0e5b5d4f81aed7ca30ab271ff","project":"openstack/zun-ui","branch":"stable/train","topic":"bug/1847227-stable/train","hashtags":[],"change_id":"Ic3dd0f949c3937f0e5b5d4f81aed7ca30ab271ff","subject":"Use labels to identify the cloud-shell container","status":"MERGED","created":"2019-10-19 22:31:59.000000000","updated":"2019-10-27 00:23:04.000000000","submitted":"2019-10-27 00:22:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689639-1572135732657-f546210b","meta_rev_id":"382ccbec1bbc8d228a8638599fced97baef31401","_number":689639,"virtual_id_number":689639,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-27 00:22:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-10-26 18:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-10-26 18:47:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-27 00:22:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b46a829a43371d619ad4630df3a21d9e4cd6986d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-19 22:31:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1d315ac7ee4c27a3eceaa91054156ff049aa8690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 22:39:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/fb28adb6c36b47b79f8e4216c7d67e6e : SUCCESS in 4m 58s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/126a76c158cd4159800ba13244be624c : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75a86e4d97424c618400bc792c54994a : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9873e3509244604996280ee70f978cc : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3bc4c79d3ee84f5b888894e44fd11034 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/944d7c64d6be440196499d2745ab0237 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46a240e3f2a04bc5b9d954d7b8aa848f : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfdc8510da024befae0e066dda518e04 : SUCCESS in 3m 34s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/2fb136ace4c44ac894570a8d523a405e : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c5051669e3396216592dbc7b65a6b1429aa48d22","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-26 18:47:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5ec1ff5d0dc757f38f3c571303799e66210d208","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 18:47:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8a035a81c7be98353e6f6924f365be9baed8a93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 18:53:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/db8cf6c3580140bca62748cc81738fe9 : SUCCESS in 4m 44s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/66a4593152ce42e4ad36620807b35b21 : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c738a266e0a405ebb60ebec47b84322 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02607d835ecb456ab34bfe958a8b776d : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a27bd3db06b483ca4f4dd7f38c98958 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e7f07f02ac24f778c7110b0f7eb8cb1 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59ff8837d579472f94902efbe8fc3750 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abfed232826944c3a206ffa2af0832ad : FAILURE in 3m 06s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/c3e93e8d4bcc46da921dd388246fb3b1 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"259cc6de92dbb08d3691aabd0101e0b2ffcc1c05","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-26 21:24:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"332b605221af7f49ccd09009d4eb121e69a8685a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 21:31:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/67064e3143804530b49662bdd193f6f8 : SUCCESS in 5m 12s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/ccb6d25b81e74133a314ed38d6c9881b : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/959454e8baa040ecab687674d13c3419 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da233d5686a94fffb5e0563cf46311ac : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18b42394631f44da8e778a698da94a4e : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/803dc95d7e7a437585a355ad38a24187 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69d289e951bc4ec38da06e8fb15a6a44 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87cf36219209449289edd5c3b385e04c : FAILURE in 3m 37s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/ed8222916fb64f85b2de9a1addbf36ec : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8e44e1f95b43db83a833617ad758ea4fb46cc2d8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-26 23:55:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1939bc42109f1460fd72aee938dcca6ef2f09e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 00:13:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/1fa952dce22748d6b71b9e34761786d4 : SUCCESS in 6m 33s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/b1d54936d2954396b81e1cd5ea0d90d8 : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c099621be2e84be2902bf10ad186d524 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd381e068064e9da8a4e3d04aae1de5 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7a8100484864cc9a8da954b061b1b70 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a67e99326c34d94a8f4f0f657678eb5 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8291f10fb6544346b2c0e44a385886ac : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70bec2f9dd00468d937664b7ea7e6cfb : SUCCESS in 3m 15s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/e29d20154150439d8259e2f6f51d2961 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"32530f89201bd59b40b0384fe4d506bf57e58dca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 00:13:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbc25284d26e4b49279c9496054d8bd54fb31359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 00:22:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/2bce050a23ca4507950789745b76ef0f : SUCCESS in 5m 01s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/d1a4ac4d8c3c438d84221e2a614c07ef : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/564e6be486674d70a71fdd23f3999164 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c8188889940455ebf1466db19f8d1b5 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a41609327c604e55b626b4770df81398 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb2a845dba98410bb2ee7a6cf9fdae26 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b94a33f309ef4664a35167386c83afef : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6db754ade7a4c87a909c45786f50c7b : SUCCESS in 3m 24s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/0f05060c25294a749496e66b00ed7553 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"251854653e887ffd7845c39a4a88fa246ab032d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 00:22:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"34d0dbd8dbbec15f28b5210d0099597a7f284c5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 00:23:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93363aee257b4a3396ba36d0949fb453 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"370a5847229605b1f03ac61caaf27e9c9b7fb5d2","revisions":{"370a5847229605b1f03ac61caaf27e9c9b7fb5d2":{"kind":"REWORK","_number":1,"created":"2019-10-19 22:31:59.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/39/689639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-ui","ref":"refs/changes/39/689639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/39/689639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/39/689639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/39/689639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-ui refs/changes/39/689639/1"}}},"commit":{"parents":[{"commit":"16fa29602d7fc0735504a2d1779ab7bbd59ecb18","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/16fa29602d7fc0735504a2d1779ab7bbd59ecb18"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-13 18:49:10.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-19 22:31:59.000000000","tz":0},"subject":"Use labels to identify the cloud-shell container","message":"Use labels to identify the cloud-shell container\n\nChange-Id: Ic3dd0f949c3937f0e5b5d4f81aed7ca30ab271ff\nCloses-Bug: #1847227\n(cherry picked from commit 2583d574d278b6bb32f7a3de1c6c1e3bdd0e8869)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/370a5847229605b1f03ac61caaf27e9c9b7fb5d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/370a5847229605b1f03ac61caaf27e9c9b7fb5d2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
