)]}'
{"id":"openstack%2Fzun-ui~691606","triplet_id":"openstack%2Fzun-ui~master~Id6674af1b9fb053e90f2b2a059517aad2ca11c67","project":"openstack/zun-ui","branch":"master","hashtags":[],"change_id":"Id6674af1b9fb053e90f2b2a059517aad2ca11c67","subject":"Stop testing python2.7","status":"ABANDONED","created":"2019-10-28 06:40:35.000000000","updated":"2020-03-30 12:20:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4d3e5641f099d8f3f3366e1bc051757dce92d3ce","_number":691606,"virtual_id_number":691606,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-28 06:59:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-2,"date":"2019-10-30 10:10:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 06:59:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-30 10:10:02.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"8bc69608073777502d2cb0c486e7be02efeaa30e","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-10-28 06:40:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"638423aae7161acbca2a19fa59f3cf26bad19172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 06:59:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/db05ca4b4b1e47b3b37ae59934ae5b4d : SUCCESS in 2m 31s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/48e12f1134ec4c0c9c1799efb8e2ef94 : SUCCESS in 4m 08s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/a57b67427a7d4759a7d0b353ee87a02b : SUCCESS in 8m 31s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/2529951a04a9476a95cdf89101a3ebc1 : SUCCESS in 6m 17s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/a043cdd7a71d461b8b74f88ad60d5871 : SUCCESS in 13m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff03949b553e48eb9fae2b50511b691c : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/472f3e6565de43babe34ce0db7cf025f : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ce94b61afa64f0a86cb0d509ea6f401 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6a5df2290bd4d6f9e97c6cd170d639a : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d3e466a7ba044d786c5fd801dba7381 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e02d5863c75499cb5927045feb68d3d : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"02acfa19df0b76c942f96e73550771457825bd15","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-10-30 10:10:02.000000000","message":"Patch Set 1: Code-Review-2\n\nThanks for the patch but the same changes are purposed in another patch[1]  in a better way.\n[1] https://review.opendev.org/#/c/692008/","accounts_in_message":[],"_revision_number":1},{"id":"4d3e5641f099d8f3f3366e1bc051757dce92d3ce","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-03-30 12:20:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3745129091fd219d2db57defae6d18dbe8266863","revisions":{"3745129091fd219d2db57defae6d18dbe8266863":{"kind":"REWORK","_number":1,"created":"2019-10-28 06:40:35.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/06/691606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-ui","ref":"refs/changes/06/691606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/06/691606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/06/691606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/06/691606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-ui refs/changes/06/691606/1"}}},"commit":{"parents":[{"commit":"49f28303ebb93b1fad45c3a135040c93cbf48008","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/49f28303ebb93b1fad45c3a135040c93cbf48008"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-28 06:39:45.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-28 06:40:24.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nChange-Id: Id6674af1b9fb053e90f2b2a059517aad2ca11c67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/3745129091fd219d2db57defae6d18dbe8266863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/3745129091fd219d2db57defae6d18dbe8266863"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
