)]}'
{"id":"openstack%2Fzun-ui~720374","triplet_id":"openstack%2Fzun-ui~master~I402993ccd928f2c845651f76d3cdf93f198d1957","project":"openstack/zun-ui","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I402993ccd928f2c845651f76d3cdf93f198d1957","subject":"Imported Translations from Zanata","status":"MERGED","created":"2020-04-16 07:29:16.000000000","updated":"2020-04-16 09:21:25.000000000","submitted":"2020-04-16 09:18:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720374-1587028728137-332a2ac9","meta_rev_id":"79ea048c3ffa1a53014761e85edf62321bed4305","_number":720374,"virtual_id_number":720374,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 09:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":2,"date":"2020-04-16 09:06:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":1,"date":"2020-04-16 09:06:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 09:06:47.000000000","updated_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"reviewer":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"state":"REVIEWER"},{"updated":"2020-04-16 09:18:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b19192934de9cfea00238fd3f0fd9fb280250598","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-16 07:29:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f82970431b73c90ff8c040dbd87147a5a960665f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 07:45:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/38b02a7a8a674ba0b23f89680dab3ad3 : SUCCESS in 8m 18s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/c92d4f3c81914ec2b87fbab7b35579f4 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fca2003533542fb8d6d999b58744774 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164890f621f54cc7a9ca79426c7d3426 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfdebac2d447457993b5f2bd7e30303e : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"43c5403bcf51b870d6a57ee931d819591a1c34ba","author":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"date":"2020-04-16 09:06:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35a4fa7d225b35028fbf7293f8d5bf92b42088f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 09:06:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f4cef182213a71d7287fe7caf157c31c932acc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 09:18:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/ed5726261d5e4411a96d6ae90ce94a24 : SUCCESS in 7m 07s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/5084769fe65d4737acd4c92961438c40 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd510d5ed1447e9af60b13d5c2d4e22 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f683727c17bd4320bf5a547a6101d672 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80814612092a4847bca8a26e3948383d : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8f9e08f75c885494da7e4eb3fbf3bbc803a405a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 09:18:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b59390aab76ea8b1969b8a520ca7fb6efc564eaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 09:21:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f22f714bccf4e5abfb6c1feff93333b : SUCCESS in 2m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89391324732b4b739101f05a2a2d5770 : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08e0830f356f80aa4e0797aa877943cd319fae5c","revisions":{"08e0830f356f80aa4e0797aa877943cd319fae5c":{"kind":"REWORK","_number":1,"created":"2020-04-16 07:29:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/720374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-ui","ref":"refs/changes/74/720374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/74/720374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/74/720374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/74/720374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-ui refs/changes/74/720374/1"}}},"commit":{"parents":[{"commit":"f3bfe42cdccb49a42ceb9143f2c20f1cf6eb3672","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/f3bfe42cdccb49a42ceb9143f2c20f1cf6eb3672"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-16 07:29:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-16 07:29:14.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I402993ccd928f2c845651f76d3cdf93f198d1957\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/08e0830f356f80aa4e0797aa877943cd319fae5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/08e0830f356f80aa4e0797aa877943cd319fae5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
