)]}'
{"id":"openstack%2Fzun-ui~809073","triplet_id":"openstack%2Fzun-ui~stable%2Fxena~I83f66baa5b645256ce81c4076d49d10c687d9f85","project":"openstack/zun-ui","branch":"stable/xena","topic":"create-xena","hashtags":[],"change_id":"I83f66baa5b645256ce81c4076d49d10c687d9f85","subject":"Update .gitreview for stable/xena","status":"MERGED","created":"2021-09-15 02:11:10.000000000","updated":"2021-09-15 10:29:09.000000000","submitted":"2021-09-15 10:27:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809073-create-xena","meta_rev_id":"aeab9ceaf41212f5ab430db04e9ecba4e4c3b23a","_number":809073,"virtual_id_number":809073,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 10:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":2,"date":"2021-09-15 10:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":1,"date":"2021-09-15 10:16:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 02:43:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 10:16:02.000000000","updated_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"reviewer":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"state":"REVIEWER"}],"messages":[{"id":"fc16960d8db8df7c442679d9a75a5c504f1602e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-15 02:11:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33300081bb4bfa92c603542800251320a6b2cc40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 02:43:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f32d6f9237d547a68d496d2027dfb78f : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2a9e5ec4ef4a20892049b01d87ae26 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01b6b685ee244a4a95c11d98e86a06f8 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9b57974d5d14c199964cdfac6486d9d : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6134b8d422bc4c2c8c5a4507eb724d9e : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a40b800e9843f59fb07b46bd353b9f : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f26783f5d2f50b07c3eebac013cd42b9a710acbb","author":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"date":"2021-09-15 10:16:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32f0767ee2a6885b53302decf99f6483010bfab2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 10:16:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4fe801fdd5f84f6dcb7695d5054fe45adca5c176","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 10:27:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1ec9149a7ff4d25a07a119407dbcb2c : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348f154a1b67427ca022fcdbffb24087 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b61871d547d8443d8597e52c185c8688 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf1304f62e9a490b807dbc4284b601c8 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a63640a216d437fa7a86253fa2a187e : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8821f9edcadffe56b77906f94295f884722e2c17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 10:27:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aeab9ceaf41212f5ab430db04e9ecba4e4c3b23a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 10:29:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e57658bf2b418db056bad4cd9ff674 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2113dc2f1f856878d58b4c1e7c0093071231c384","revisions":{"2113dc2f1f856878d58b4c1e7c0093071231c384":{"kind":"REWORK","_number":1,"created":"2021-09-15 02:11:10.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/73/809073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-ui","ref":"refs/changes/73/809073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/73/809073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/73/809073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/73/809073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-ui refs/changes/73/809073/1"}}},"commit":{"parents":[{"commit":"f27296e6d5605036c773bd947d646feadc30d919","subject":"Drop horizon-nodejs10-jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/f27296e6d5605036c773bd947d646feadc30d919"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-15 02:11:09.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-15 02:11:09.000000000","tz":0},"subject":"Update .gitreview for stable/xena","message":"Update .gitreview for stable/xena\n\nChange-Id: I83f66baa5b645256ce81c4076d49d10c687d9f85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/2113dc2f1f856878d58b4c1e7c0093071231c384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/2113dc2f1f856878d58b4c1e7c0093071231c384"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
