)]}'
{"id":"openstack%2Fzun-ui~809076","triplet_id":"openstack%2Fzun-ui~master~Id6871b692490abc6e4d66d02e95f6e65eb41cd7f","project":"openstack/zun-ui","branch":"master","topic":"add-yoga-python-jobtemplates","hashtags":[],"change_id":"Id6871b692490abc6e4d66d02e95f6e65eb41cd7f","subject":"Add Python3 yoga unit tests","status":"MERGED","created":"2021-09-15 02:11:14.000000000","updated":"2021-09-15 23:32:34.000000000","submitted":"2021-09-15 23:28:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809076-add-yoga-python-jobtemplates","meta_rev_id":"2100019044a83371b27280b275a51b49a23dd657","_number":809076,"virtual_id_number":809076,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 23:28:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2021-09-15 12:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2021-09-15 23:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2021-09-15 23:17:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 02:56:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 12:57:29.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2021-09-15 23:17:54.000000000","updated_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"reviewer":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"state":"REVIEWER"}],"messages":[{"id":"2573cbe9cfc18005dd1fe4b7315df9f7f2d83904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-15 02:11:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0987161f41e18b1ce64150a0e5622618310c8623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 02:56:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5a4143d1f6648388368853c87817fac : SUCCESS in 6m 56s\n- horizon-tox-python3-django30 https://zuul.opendev.org/t/openstack/build/dca301b67af64a0db73bbb9574cf0384 : SUCCESS in 8m 01s (non-voting)\n- horizon-tox-python3-django31 https://zuul.opendev.org/t/openstack/build/66f9156a443349fdb476ae5a1b7c7407 : SUCCESS in 7m 48s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/c1da06540c1244898d31847e01aa55d3 : SUCCESS in 7m 48s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/8fd273a938e84523b305f5e2bdf48993 : TIMED_OUT in 32m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89e1d75213a847a09e6a30a87f93acaf : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642d2b22e43d4c408fb46543171a3ab7 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f2d9b9125594309889368e314e9452f : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9779b15f9e44f438e98d962cbb5807c : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1634c50689664a8e9511fc2cf4646fea : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":1},{"id":"291377acb575997bdd69e5a451de509d1e8db29d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-09-15 12:57:29.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"03ed7c5a6d71b6a8e64c24c8ec630135e46e1f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 13:13:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d5a84b13a98492d825a6e1e20b87889 : SUCCESS in 5m 26s\n- horizon-tox-python3-django30 https://zuul.opendev.org/t/openstack/build/54f0da758aa04800a96b15a15704e260 : SUCCESS in 6m 13s (non-voting)\n- horizon-tox-python3-django31 https://zuul.opendev.org/t/openstack/build/37f21ef6d427454e870a2f7cb9bb9548 : SUCCESS in 5m 18s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/5c64f2f4d7ed46b0b3be0fae3cf40624 : SUCCESS in 5m 40s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/6719f172a81d48d6bf0df4dc7a8ae27e : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec96f7259079499984022b9359ee8335 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5013c9732d340d6aff7719f188d23b3 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc61a935df854463b93e9335278fe7cd : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d38d48a2b3194530ac91796e5be8d2ef : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3d756c0b3844ce9a4e4f3258f9c69f : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bfae8429ef1c3182cf329c0e3a7a9592359dd6e0","author":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"date":"2021-09-15 23:17:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c1f93f73990b521209706caa0b244733ff0ac63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 23:18:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1865ee0d8dc4133546f7af2c73eb7387529eaf63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 23:28:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3920aa172e04a12973c5f0f7d39806e : SUCCESS in 6m 08s\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/e9cad7db61074688b4ab3e6a2b710418 : SUCCESS in 6m 19s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/e6eb7398422d41f983bd621ffa44ada5 : SUCCESS in 8m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10ba459b3e2b453983a116c7a6e88731 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/069aba0492c1448a89dc1dd6f18fd63c : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2159c5acd374e4888908cf4b5d70b2c : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7204df4d1a7d43d7ac0c55ebab3f4f2c : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":1},{"id":"708cb115c872d996610763857dd7ad954fcef29d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 23:28:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2100019044a83371b27280b275a51b49a23dd657","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 23:32:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2991eee7e9a94db997369bd566366bdb : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"267cf063332a338f97c490a9def76946c88dd194","revisions":{"267cf063332a338f97c490a9def76946c88dd194":{"kind":"REWORK","_number":1,"created":"2021-09-15 02:11:14.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/76/809076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-ui","ref":"refs/changes/76/809076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/76/809076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/76/809076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-ui refs/changes/76/809076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-ui refs/changes/76/809076/1"}}},"commit":{"parents":[{"commit":"f3487db23fa55114a97bb154728a76c71a41070a","subject":"Update master for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/f3487db23fa55114a97bb154728a76c71a41070a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-15 02:11:13.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-15 02:11:13.000000000","tz":0},"subject":"Add Python3 yoga unit tests","message":"Add Python3 yoga unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for yoga.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Id6871b692490abc6e4d66d02e95f6e65eb41cd7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/267cf063332a338f97c490a9def76946c88dd194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-ui/commit/267cf063332a338f97c490a9def76946c88dd194"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
