)]}'
{"id":"openstack%2Fzun~1002921","triplet_id":"openstack%2Fzun~master~I759c76264d8b5fc7fecf04be3593b93d4476c952","project":"openstack/zun","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-30 10:25:14.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I759c76264d8b5fc7fecf04be3593b93d4476c952","subject":"bindep: Remove stale markers","status":"MERGED","created":"2026-08-30 01:45:35.000000000","updated":"2026-08-30 12:13:29.000000000","submitted":"2026-08-30 12:12:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002921","meta_rev_id":"e75c500c78575ace679f0f04c72e0ce403dcdb01","_number":1002921,"virtual_id_number":1002921,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-30 12:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-30 11:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-30 11:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-30 03:54:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52e48fc5ec4f6f6900e6db16de01be80ea9f5255","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 01:45:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f32763a47eb1865de9a6b222d9546231e8d14dd9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 01:46:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f924f41235067d531670d608ddb5c06f984c81df","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 01:52:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d170beb89fac11b2d6cfb1a8d669b6c470819ce6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:02:01.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c46781ac90f5b0c0321e69dfbf785945a78e2027","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:02:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"5a2ec2cbbbb19da603c8601e4834bd62a5529658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 03:54:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c50b588953e450f87089483e1e996c5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d0bb98263bd74a81a6ee19845d90102e : SUCCESS in 4m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6840480c4cd746129d84102cea49bccd : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cb52bd83f994c478028a0afea4e816a : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df7c3aae44a84586a03a695018e7b10d : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18ee8a95efa44568ae6b7e56bdbc0236 : SUCCESS in 6m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eb3c3d1dcf93441a9e1cbfe0eb982299 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0848d762db4060925eec559b21a8ae : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbe1fdd7d5354b1a9db9562d0232ab9f : SUCCESS in 8m 52s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/bcb52962cb2c4a218100d75440bb1a7b : FAILURE in 28m 25s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/fb5431202849415db259a2ba64f1fcfc : SUCCESS in 48m 12s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/38bffedbafe8424596cadac3b5648d12 : SUCCESS in 24m 19s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/d7392a64a49d405a9e63317d85239f57 : FAILURE in 27m 30s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/3809f73ef5104d2882c2f18204f787c5 : FAILURE in 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b1992b71bbd72ddbdf9251038781381f3d86b8a2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 10:25:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d38a7c2362b4f5c39b3ba1ab84669d80b563bcb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 11:23:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa642dbd2f7b4fadbbaefd446d6f5303\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3cc83113104d48c388c44e395cd38a11 : SUCCESS in 4m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ba302b4621347cb9f17dc4f88f724ba : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be914aa68ec14c4888e363b6a73d5658 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e5c339c426847e9878825e59db623b6 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fbe1b46c93949e0ae52061a54c03f3b : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c0f555d2f27a4558af583b86aed2df89 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e6ff99d74a4ebbb044d4b2b605c4a7 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/405a4d05a6e44f32888c425efd306105 : SUCCESS in 6m 47s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/09b9e2bae9b240e5bcc31e2d5c1d1109 : SUCCESS in 48m 10s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/3e37862b2dc84f27a784619c28eb104b : SUCCESS in 29m 37s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/f70ddeb45f00463ca4b763bbea1f01a3 : SUCCESS in 33m 15s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/db593eb1d29b4b7f95e06af1f2dd0b3c : FAILURE in 18m 51s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/2ce50a98c5f04443aca4dece721ff591 : SUCCESS in 23m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f0c4ffaaa09b2088a5f3dfcd3913faa275388820","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 11:25:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8548f57df08880f76e1e38fbe49fc5f92272698f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 11:25:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fcd9bc3f7c0f2fa5a1c0e6b3b633ce24a371facd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 12:12:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ee9b721bc244d629e1d4a55e241dfb5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/37914316d3be4ba58cb67822bbcc275b : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb13ce239eec4a588b62aa7848193739 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f7e14ad9a2042ff9690b0b187b543d9 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e459e445e23d45f39df894db77fb05d8 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06723823b76749c79ea5a0b783da1927 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/835cf873ea77438eac8ea423445fa491 : SUCCESS in 5m 07s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/10133ce51f6440cf9091524db5ba7361 : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":4},{"id":"e2d65f3fa31065a9c5f297e239558bf455ace0ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 12:12:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e75c500c78575ace679f0f04c72e0ce403dcdb01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 12:13:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ff1f5fbf22a4398bbabaf712dffe769\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/fe5cecb6165c47a3b7ae6e5a180d8c65 : SUCCESS in 52s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f597b47d713e48e681c8eab31c714e65 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d22108932b5545f9a57555bb0c11ab3e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b191ed7493c4d4be9e5660b3a050f3833eb923a6","revisions":{"0308244f3646fbc916d797ffaa75046d7e35112c":{"kind":"REWORK","_number":1,"created":"2026-08-30 01:45:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/1002921/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/21/1002921/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/21/1002921/1"}}},"commit":{"parents":[{"commit":"7ce591604bea3e174abdff4918515173d669e860","subject":"Merge \"Remove transactional.SYNC_TIMEOUT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7ce591604bea3e174abdff4918515173d669e860"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 01:45:35.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 01:45:35.000000000","tz":0},"subject":"bindep: Remove stale markers","message":"bindep: Remove stale markers\n\nRemove markers for old operating system versions which are no longer\nsupported.\n\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I759c76264d8b5fc7fecf04be3593b93d4476c952\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0308244f3646fbc916d797ffaa75046d7e35112c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0308244f3646fbc916d797ffaa75046d7e35112c"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"05928fb38a6c35ddcde805f0f068f4abbd2a23a6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-30 01:46:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/1002921/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/21/1002921/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/21/1002921/2"}}},"commit":{"parents":[{"commit":"7ce591604bea3e174abdff4918515173d669e860","subject":"Merge \"Remove transactional.SYNC_TIMEOUT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7ce591604bea3e174abdff4918515173d669e860"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 01:45:35.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 01:46:00.000000000","tz":0},"subject":"bindep: Remove stale markers","message":"bindep: Remove stale markers\n\nRemove markers for old operating system versions which are no longer\nsupported.\n\nChange-Id: I759c76264d8b5fc7fecf04be3593b93d4476c952\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/05928fb38a6c35ddcde805f0f068f4abbd2a23a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/05928fb38a6c35ddcde805f0f068f4abbd2a23a6"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"5bf556c29deda11b577ebdbdfcc255a03e63500d":{"kind":"REWORK","_number":3,"created":"2026-08-30 01:52:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/1002921/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/21/1002921/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/21/1002921/3"}}},"commit":{"parents":[{"commit":"7ce591604bea3e174abdff4918515173d669e860","subject":"Merge \"Remove transactional.SYNC_TIMEOUT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7ce591604bea3e174abdff4918515173d669e860"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 01:45:35.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 01:52:00.000000000","tz":0},"subject":"bindep: Remove stale markers","message":"bindep: Remove stale markers\n\nRemove markers for old operating system versions which are no longer\nsupported.\n\nChange-Id: I759c76264d8b5fc7fecf04be3593b93d4476c952\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5bf556c29deda11b577ebdbdfcc255a03e63500d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5bf556c29deda11b577ebdbdfcc255a03e63500d"}]},"branch":"refs/heads/master"},"b191ed7493c4d4be9e5660b3a050f3833eb923a6":{"kind":"REWORK","_number":4,"created":"2026-08-30 02:02:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/21/1002921/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/21/1002921/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/1002921/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/21/1002921/4"}}},"commit":{"parents":[{"commit":"7ce591604bea3e174abdff4918515173d669e860","subject":"Merge \"Remove transactional.SYNC_TIMEOUT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7ce591604bea3e174abdff4918515173d669e860"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 01:45:35.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:02:01.000000000","tz":0},"subject":"bindep: Remove stale markers","message":"bindep: Remove stale markers\n\nRemove markers for old operating system versions which are no longer\nsupported.\n\nChange-Id: I759c76264d8b5fc7fecf04be3593b93d4476c952\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b191ed7493c4d4be9e5660b3a050f3833eb923a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b191ed7493c4d4be9e5660b3a050f3833eb923a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
