)]}'
{"id":"openstack%2Fzun~1005849","triplet_id":"openstack%2Fzun~master~I66cff26555d24276d872679cad7e83e6c3fc9604","project":"openstack/zun","branch":"master","attention_set":{},"removed_from_attention_set":{"13039":{"account":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"last_update":"2026-09-17 16:53:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66cff26555d24276d872679cad7e83e6c3fc9604","subject":"Set shareable IPC mode for capsule sandbox","status":"MERGED","created":"2026-09-16 09:40:23.000000000","updated":"2026-09-17 16:55:13.000000000","submitted":"2026-09-17 16:53:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005849","meta_rev_id":"e9a5988f1c8ab9fff26a99558b4ada1aef3297c7","_number":1005849,"virtual_id_number":1005849,"owner":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 16:53:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-17 16:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-17 16:01:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 10:22:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 16:01:22.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"fc13987dacf99490ebbdeb9c0d3dc45fb49ec5be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2026-09-16 09:40:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ce1fdb040e4a179187f7e1a9dde3f738102be6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 10:22:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84fa850220864f0f9e44fb3200f0a4e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37d45118047a494894fc323dfe6744ab : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/946603b67e7742049aacf43b697c6a62 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/631306c7f8bc4cc3b543022c96783f1e : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce2b6e54e25749c0bef7dc2b3b6a17e0 : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aeaeae0745cf4ea690c75c55a2cb52bb : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1989a7dab454c77ad58657f844374bf : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee82982eac6449228234562f58bfdaf8 : SUCCESS in 6m 49s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/dc8ec5032e954c85b7c6e9c06fe21690 : SUCCESS in 37m 06s\n- zun-tempest-docker-sql-ipv6-only https://zuul.opendev.org/t/openstack/build/45e0a7c811254e17893398794d557c84 : SUCCESS in 23m 35s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2ea07f74ad28404d8c1925ef47d78887 : SUCCESS in 39m 24s (non-voting)\n- zun-fullstack https://zuul.opendev.org/t/openstack/build/986f4c6f43a24efe9a02c69dfa369c78 : FAILURE in 31m 08s (non-voting)\n- zun-tempest-docker-sql-standalone https://zuul.opendev.org/t/openstack/build/802debaf604c4a8db58b96bee6543508 : SUCCESS in 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eabc291f5b4bec34bb0b0b5cafef55738bcde7a0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-17 16:01:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0fdf9cccc076a347a893f76c4f729b39bf09bb8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:02:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81dfe0e66a30cf62254b82db368c43506ddf3d12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:53:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/052b63b287c44a9291116d097a5dd0cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fafa289614c443ec97afa2d01710d0e3 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45e27bf48a31448f9f5dc7bc9998bed7 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03bcbae3096245cbb4fe667c7bdcce06 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc23a6c6fb824da88c9774eff47835ca : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a5366db3a024b9497c6ee062ce33637 : SUCCESS in 8m 02s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9c0147d8f6424b6bb5ba980b13659395 : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d668c4759349a0242cc97650a25b5b9af48bb266","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:53:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e9a5988f1c8ab9fff26a99558b4ada1aef3297c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:55:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b94f4f2e16014b04be63b3d6f8eb4380\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091d1baf1b274a09b46e1498629558af : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d33e1c552deb6c124086f3a46b41cd2a5816c9e","revisions":{"6d33e1c552deb6c124086f3a46b41cd2a5816c9e":{"kind":"REWORK","_number":1,"created":"2026-09-16 09:40:23.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/49/1005849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/49/1005849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/1005849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/1005849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/49/1005849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/49/1005849/1"}}},"commit":{"parents":[{"commit":"26e07b20f2cca3fb513abbaf02569e5fe60e5c2e","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/26e07b20f2cca3fb513abbaf02569e5fe60e5c2e"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2026-09-16 09:39:02.000000000","tz":120},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2026-09-16 09:40:13.000000000","tz":120},"subject":"Set shareable IPC mode for capsule sandbox","message":"Set shareable IPC mode for capsule sandbox\n\nWhen creating a capsule\u0027s sandbox container, set\nipc_mode\u003d\u0027shareable\u0027 in the host config so that child\ncontainers can join its IPC namespace.\n\nDocker API \u003e\u003d 1.30 defaults containers to IpcMode:private,\nwhich broke capsule creation after commit 072e4d8e bumped\ndocker_remote_api_version from 1.26 to 1.40. Child\ncontainers that specify ipc_mode\u003d\u0027container:\u003csandbox_id\u003e\u0027\nnow fail with \u0027non-shareable IPC\u0027.\n\nCloses-Bug: #2167446\n\nChange-Id: I66cff26555d24276d872679cad7e83e6c3fc9604\nSigned-off-by: Martin André \u003cm.andre@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6d33e1c552deb6c124086f3a46b41cd2a5816c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6d33e1c552deb6c124086f3a46b41cd2a5816c9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
