)]}'
{"id":"openstack%2Fzun~652240","triplet_id":"openstack%2Fzun~master~I9945087cf8d1484c09e60d11b1ba0c84c1799f00","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I9945087cf8d1484c09e60d11b1ba0c84c1799f00","subject":"Record action finish","status":"MERGED","created":"2019-04-13 23:09:36.000000000","updated":"2019-05-27 03:30:05.000000000","submitted":"2019-05-27 03:30:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652240-1558927805512-33c96c3c","meta_rev_id":"b3fcbbc247f74af2cca33ffb4e201c98c84acac6","_number":652240,"virtual_id_number":652240,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-05-27 03:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-05-27 02:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-05-27 02:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 12:16:52.000000000","updated_by":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"reviewer":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"state":"REVIEWER"},{"updated":"2019-05-27 02:26:11.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-05-27 03:30:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b7f9c19730514b949d77474737302445aa8afc9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-13 23:09:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"252f43d5505e653b62910c054a380a2e49093f6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-14 01:04:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/1/check/openstack-tox-cover/b875e5a/ : FAILURE in 4m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/1/check/openstack-tox-lower-constraints/9c908d1/ : FAILURE in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/1/check/openstack-tox-pep8/acbd4cf/ : FAILURE in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/1/check/openstack-tox-py27/7fe9bd1/ : FAILURE in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/40/652240/1/check/openstack-tox-py35/5774051/ : FAILURE in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/1/check/openstack-tox-py36/ee3f53e/ : FAILURE in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/1/check/openstack-tox-py37/c3fd43f/ : FAILURE in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/40/652240/1/check/openstack-tox-docs/01afa3f/html/ : SUCCESS in 3m 43s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/1/check/zun-tempest-docker-sql/be6e694/ : TIMED_OUT in 1h 13m 47s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/40/652240/1/check/zun-tempest-py35-docker-sql/f3c720c/ : TIMED_OUT in 1h 13m 28s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/1/check/zun-tempest-docker-etcd/6874fc3/ : TIMED_OUT in 1h 13m 20s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/1/check/zun-tempest-multinode-docker-sql/a64638e/ : FAILURE in 1h 53m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3cfea5897a4a8f5f06e499bb264a68b6e3ffe0f6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-14 16:04:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"723a93a25eb8ee3169dc9b9d5b137e3671f534b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-14 17:02:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/2/check/openstack-tox-cover/61706c0/ : FAILURE in 4m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/2/check/openstack-tox-lower-constraints/63ea85a/ : FAILURE in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/2/check/openstack-tox-pep8/1a7a02d/ : FAILURE in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/2/check/openstack-tox-py27/c86a66a/ : FAILURE in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/40/652240/2/check/openstack-tox-py35/11601b6/ : FAILURE in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/2/check/openstack-tox-py36/9fe2813/ : FAILURE in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/2/check/openstack-tox-py37/1b79710/ : FAILURE in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/40/652240/2/check/openstack-tox-docs/b48d53f/html/ : SUCCESS in 3m 33s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/2/check/zun-tempest-docker-sql/4351ad4/ : SUCCESS in 46m 17s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/40/652240/2/check/zun-tempest-py35-docker-sql/22b127a/ : SUCCESS in 45m 40s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/2/check/zun-tempest-docker-etcd/ba6996c/ : SUCCESS in 44m 37s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/2/check/zun-tempest-multinode-docker-sql/eb3d91b/ : SUCCESS in 56m 39s","accounts_in_message":[],"_revision_number":2},{"id":"6738bc33c3aba0736286d7f4934c47f7b0959745","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-14 19:01:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"49f600647610bd5311b6b41cb085c84ad6a5bab2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-14 19:53:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/3/check/openstack-tox-cover/9437a2d/cover/ : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/3/check/openstack-tox-lower-constraints/4c56d63/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/3/check/openstack-tox-pep8/0956ad0/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/3/check/openstack-tox-py27/5621236/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/40/652240/3/check/openstack-tox-py35/32f3574/ : SUCCESS in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/3/check/openstack-tox-py36/916b4aa/ : SUCCESS in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/3/check/openstack-tox-py37/164b1e2/ : SUCCESS in 5m 40s\n- openstack-tox-docs http://logs.openstack.org/40/652240/3/check/openstack-tox-docs/c07803d/html/ : SUCCESS in 4m 01s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/3/check/zun-tempest-docker-sql/1d3dd03/ : SUCCESS in 46m 34s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/40/652240/3/check/zun-tempest-py35-docker-sql/af5a7d5/ : SUCCESS in 39m 41s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/3/check/zun-tempest-docker-etcd/ceac251/ : SUCCESS in 50m 03s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/3/check/zun-tempest-multinode-docker-sql/69e77dc/ : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":3},{"id":"5a9296a12bf068efb889ffa20f8630bed979ab74","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-14 21:53:47.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"96ec9a3526d669354c0111e43266094889fb748a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-14 22:59:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/4/check/openstack-tox-cover/1c44a5f/cover/ : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/4/check/openstack-tox-lower-constraints/28bbeb7/ : SUCCESS in 5m 32s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/4/check/openstack-tox-pep8/d0871f6/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/4/check/openstack-tox-py27/79cbaa0/ : SUCCESS in 4m 22s\n- openstack-tox-py35 http://logs.openstack.org/40/652240/4/check/openstack-tox-py35/187bf3f/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/4/check/openstack-tox-py36/842108d/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/4/check/openstack-tox-py37/f69431c/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/40/652240/4/check/openstack-tox-docs/7322b34/html/ : SUCCESS in 3m 39s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/4/check/zun-tempest-docker-sql/cb179cf/ : SUCCESS in 50m 44s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/40/652240/4/check/zun-tempest-py35-docker-sql/6e2b4e2/ : SUCCESS in 48m 13s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/4/check/zun-tempest-docker-etcd/bdd335e/ : SUCCESS in 49m 18s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/4/check/zun-tempest-multinode-docker-sql/241cb83/ : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":4},{"id":"f4f71385bab62339742e708571e5fa986f8728e8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-15 00:23:28.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"71e9635e1d7bc4aff286f20ebe91d2191976e08a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-15 01:14:19.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"7e617fce4da22d30e7c4a9dfb884f3b943f0b5b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 02:24:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/6/check/openstack-tox-cover/bd0506a/cover/ : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/6/check/openstack-tox-lower-constraints/273a67c/ : SUCCESS in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/6/check/openstack-tox-pep8/2e01fd8/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/6/check/openstack-tox-py27/04d008a/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/40/652240/6/check/openstack-tox-py35/40bfbca/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/6/check/openstack-tox-py36/d2406d0/ : SUCCESS in 4m 14s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/6/check/openstack-tox-py37/91c67c0/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/40/652240/6/check/openstack-tox-docs/284053e/html/ : SUCCESS in 3m 59s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/6/check/zun-tempest-docker-sql/da5d3b0/ : SUCCESS in 46m 22s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/40/652240/6/check/zun-tempest-py35-docker-sql/a500bc9/ : SUCCESS in 44m 47s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/6/check/zun-tempest-docker-etcd/56abbd3/ : SUCCESS in 46m 49s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/6/check/zun-tempest-multinode-docker-sql/5334552/ : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":6},{"id":"a4be8a58209ce1519bfd6c9f9e6d90f30c0fde84","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-17 02:23:27.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3abcf7a9f5de2ce16cd844c652a1ef8eb7d7d335","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 03:30:53.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/6/check/openstack-tox-cover/8600095/cover/ : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/6/check/openstack-tox-lower-constraints/deca634/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/6/check/openstack-tox-pep8/9165574/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/6/check/openstack-tox-py27/e0bdcd0/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/6/check/openstack-tox-py36/654a57d/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/6/check/openstack-tox-py37/5cfb5bb/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/40/652240/6/check/openstack-tox-docs/376f89f/html/ : SUCCESS in 3m 28s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/6/check/zun-tempest-docker-sql/1ebc185/ : SUCCESS in 49m 30s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/6/check/zun-tempest-py3-docker-sql/c283ccf/ : SUCCESS in 45m 39s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/6/check/zun-tempest-docker-etcd/76b69ea/ : SUCCESS in 53m 52s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/6/check/zun-tempest-multinode-docker-sql/8ce10ab/ : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":6},{"id":"f5dd3f06685970fc4d81e079c0621c1ab59d521d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-20 20:15:27.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"669170f717850b8550e9889c1bdad633eaabf098","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 21:43:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/7/check/openstack-tox-cover/bba00a2/cover/ : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/7/check/openstack-tox-lower-constraints/1540e26/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/7/check/openstack-tox-pep8/0a82cd0/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/7/check/openstack-tox-py27/05cba7c/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/7/check/openstack-tox-py36/ff15339/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/7/check/openstack-tox-py37/2839c19/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/40/652240/7/check/openstack-tox-docs/ce76dd9/html/ : SUCCESS in 3m 19s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/7/check/zun-tempest-docker-sql/511574e/ : SUCCESS in 41m 40s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/7/check/zun-tempest-py3-docker-sql/8cbe489/ : SUCCESS in 48m 18s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/7/check/zun-tempest-docker-etcd/c296582/ : SUCCESS in 47m 04s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/7/check/zun-tempest-multinode-docker-sql/e2cf986/ : SUCCESS in 1h 09m 15s","accounts_in_message":[],"_revision_number":7},{"id":"7c8bbb8b72032fb7586f7f53a8c71f84283e3e0a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-21 00:02:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"84be2e7aee27ce982ea8bc26f8b5cfca7771d23f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 01:28:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/8/check/openstack-tox-cover/c055ade/cover/ : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/8/check/openstack-tox-lower-constraints/23050d1/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/8/check/openstack-tox-pep8/b48c26f/ : SUCCESS in 6m 12s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/8/check/openstack-tox-py27/52860f4/ : SUCCESS in 5m 01s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/8/check/openstack-tox-py36/3526ff9/ : SUCCESS in 5m 30s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/8/check/openstack-tox-py37/a1de2c8/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/40/652240/8/check/openstack-tox-docs/2485012/html/ : SUCCESS in 4m 06s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/8/check/zun-tempest-docker-sql/15b0455/ : SUCCESS in 52m 47s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/8/check/zun-tempest-py3-docker-sql/fe557b8/ : SUCCESS in 51m 37s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/8/check/zun-tempest-docker-etcd/fde0808/ : SUCCESS in 48m 32s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/8/check/zun-tempest-multinode-docker-sql/ab13261/ : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":8},{"id":"fcdf45eb62d98fb84dc00b606bf8654772fecbb7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-21 02:53:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"283125b1ec5967a0c69a3b132105d934052e89d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 03:58:40.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/9/check/openstack-tox-cover/65e0f79/cover/ : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/9/check/openstack-tox-lower-constraints/e87eb23/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/9/check/openstack-tox-pep8/e7b7f4d/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/9/check/openstack-tox-py27/3cdc8a7/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/9/check/openstack-tox-py36/bbbbbca/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/9/check/openstack-tox-py37/7d448fc/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/40/652240/9/check/openstack-tox-docs/1de0f96/html/ : SUCCESS in 3m 04s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/9/check/zun-tempest-docker-sql/2b0bad8/ : FAILURE in 49m 06s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/9/check/zun-tempest-py3-docker-sql/e4bcce8/ : SUCCESS in 57m 49s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/9/check/zun-tempest-docker-etcd/1904e85/ : SUCCESS in 59m 23s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/9/check/zun-tempest-multinode-docker-sql/258588b/ : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":9},{"id":"66202940defd5a368c9bddda82edebde04fd1d88","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-21 04:32:19.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"af9b75b242a602ab62fee975c2b36dd8e5e9d834","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:49:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/9/check/openstack-tox-cover/9feb80b/cover/ : SUCCESS in 7m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/9/check/openstack-tox-lower-constraints/ff3f9e2/ : SUCCESS in 5m 36s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/9/check/openstack-tox-pep8/3abba30/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/9/check/openstack-tox-py27/bb39a25/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/9/check/openstack-tox-py36/6a9b618/ : SUCCESS in 6m 12s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/9/check/openstack-tox-py37/e728545/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/40/652240/9/check/openstack-tox-docs/e732589/html/ : SUCCESS in 4m 49s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/9/check/zun-tempest-docker-sql/c387213/ : SUCCESS in 49m 09s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/9/check/zun-tempest-py3-docker-sql/6d00d77/ : SUCCESS in 52m 56s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/9/check/zun-tempest-docker-etcd/cec9e16/ : SUCCESS in 49m 44s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/9/check/zun-tempest-multinode-docker-sql/6244250/ : SUCCESS in 1h 04m 35s","accounts_in_message":[],"_revision_number":9},{"id":"20ed141a4ac5c2d038aacbab680b695a3994b333","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-24 01:18:11.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"75580e04390d49a906ec44fa738cde4acfe9b1f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 02:44:35.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/10/check/openstack-tox-cover/326a72d/cover/ : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/10/check/openstack-tox-lower-constraints/f5db268/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/10/check/openstack-tox-pep8/4b8e37b/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/10/check/openstack-tox-py27/86002d5/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/10/check/openstack-tox-py36/0cce6df/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/10/check/openstack-tox-py37/8dab923/ : SUCCESS in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/40/652240/10/check/openstack-tox-docs/ac87add/html/ : SUCCESS in 4m 10s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/10/check/zun-tempest-docker-sql/185fcdf/ : SUCCESS in 52m 55s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/10/check/zun-tempest-py3-docker-sql/4f95b2f/ : SUCCESS in 49m 16s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/10/check/zun-tempest-docker-etcd/a0682e7/ : SUCCESS in 49m 14s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/10/check/zun-tempest-multinode-docker-sql/5a2ab14/ : FAILURE in 1h 19m 52s","accounts_in_message":[],"_revision_number":10},{"id":"001e87896feaa6d196340da18404040deabe0a6a","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-04-24 10:12:26.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)\n\nHi Hongbin, I have a suggestion inline. Thanks!","accounts_in_message":[],"_revision_number":10},{"id":"1ee98ff4a4e3d49542c5b1c3780a9b09bd0698ea","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-25 02:38:07.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f298f6281305a5bc09f47eaa7e5d54ff9069d9d7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-27 02:50:14.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"20ed4af242959332c4415216524963ea276c009e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-27 02:53:13.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d718f9b1aee1ea7f5ef148569ff821dc88f8b653","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 04:07:59.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/11/check/openstack-tox-cover/800a656/cover/ : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/11/check/openstack-tox-lower-constraints/b909bb6/ : SUCCESS in 6m 45s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/11/check/openstack-tox-pep8/f88a184/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/11/check/openstack-tox-py27/307f1f2/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/11/check/openstack-tox-py36/ccf49a2/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/11/check/openstack-tox-py37/2d250b6/ : SUCCESS in 6m 10s\n- openstack-tox-docs http://logs.openstack.org/40/652240/11/check/openstack-tox-docs/c752ced/html/ : SUCCESS in 3m 29s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/11/check/zun-tempest-docker-sql/50b0e69/ : FAILURE in 45m 01s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/11/check/zun-tempest-py3-docker-sql/f17cc6e/ : SUCCESS in 48m 06s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/11/check/zun-tempest-docker-etcd/6a14175/ : SUCCESS in 57m 46s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/11/check/zun-tempest-multinode-docker-sql/26c0d4f/ : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":11},{"id":"eecebf73e6ede487fcd8812220ffae4160c98819","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-27 18:15:00.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"794d264448ddf12781e17b59d9fd36c6d439760c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-27 18:19:10.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"787e5e1f2c81756613f44a25f93bc6fea9e4c08d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 19:26:38.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/11/check/openstack-tox-cover/ddb870f/cover/ : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/11/check/openstack-tox-lower-constraints/3248a94/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/11/check/openstack-tox-pep8/d9ba1cb/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/11/check/openstack-tox-py27/bc7c5aa/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/11/check/openstack-tox-py36/e1bd0fd/ : SUCCESS in 5m 00s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/11/check/openstack-tox-py37/451d271/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/40/652240/11/check/openstack-tox-docs/46c8ab6/html/ : SUCCESS in 3m 29s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/11/check/zun-tempest-docker-sql/6828caf/ : SUCCESS in 48m 11s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/11/check/zun-tempest-py3-docker-sql/80947f8/ : SUCCESS in 47m 16s\n- zun-tempest-docker-etcd http://logs.openstack.org/40/652240/11/check/zun-tempest-docker-etcd/712aa4e/ : SUCCESS in 50m 40s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/11/check/zun-tempest-multinode-docker-sql/b936226/ : SUCCESS in 59m 12s","accounts_in_message":[],"_revision_number":11},{"id":"67bb6e30e08c137bb73314a9c3f9f0307d6ce5df","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-04-29 12:16:52.000000000","message":"Patch Set 11: Code-Review+2\n\nLGTM. Thanks!","accounts_in_message":[],"_revision_number":11},{"id":"144e2627c6f251ef98c6c4b51d52eb1b55905ef0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 16:02:50.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a343a24bd8e54f51734e4d646bb500e4925e51a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 17:00:49.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/12/check/openstack-tox-cover/551e953/cover/ : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/12/check/openstack-tox-lower-constraints/4ec29da/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/12/check/openstack-tox-pep8/9492e18/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/12/check/openstack-tox-py27/3d7bc03/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/12/check/openstack-tox-py36/1212b16/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/12/check/openstack-tox-py37/3954ef0/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/40/652240/12/check/openstack-tox-docs/cc433a7/html/ : SUCCESS in 3m 22s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/12/check/zun-tempest-docker-sql/caab3a3/ : SUCCESS in 47m 11s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/12/check/zun-tempest-py3-docker-sql/f38af48/ : SUCCESS in 41m 21s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/12/check/zun-tempest-multinode-docker-sql/88c0688/ : SUCCESS in 56m 33s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/40/652240/12/check/kolla-ansible-ubuntu-source-zun/63ab97a/ : SUCCESS in 43m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1d3048c3909ec28c281824d533781d1236176a21","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-25 03:57:50.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"50b54001b683330f2873d73acea7d39b80249f3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 04:52:04.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/40/652240/13/check/openstack-tox-cover/ba3ee65/cover/ : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/13/check/openstack-tox-lower-constraints/da8c931/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/13/check/openstack-tox-pep8/8c7d791/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/13/check/openstack-tox-py27/a11c808/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/13/check/openstack-tox-py36/b133e8d/ : SUCCESS in 4m 12s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/13/check/openstack-tox-py37/99aea54/ : SUCCESS in 4m 26s\n- openstack-tox-docs http://logs.openstack.org/40/652240/13/check/openstack-tox-docs/c057111/html/ : SUCCESS in 2m 53s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/13/check/zun-tempest-docker-sql/cc7edb8/ : SUCCESS in 48m 06s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/13/check/zun-tempest-py3-docker-sql/54fd958/ : SUCCESS in 42m 43s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/13/check/zun-tempest-multinode-docker-sql/68da6d6/ : SUCCESS in 52m 49s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/40/652240/13/check/kolla-ansible-ubuntu-source-zun/23302ba/ : SUCCESS in 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f72ef98c3f3611db2e3d6e0da742de841c666ba3","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-05-27 02:26:11.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"1a3e232aa0eabd915b7d88c80f87b4385fc8dfbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 02:26:26.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"43c10058089e201991e3cfad5b0ac95d9d239efe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 03:30:05.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/40/652240/13/gate/openstack-tox-lower-constraints/e45bfe7/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/40/652240/13/gate/openstack-tox-pep8/a59535c/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/40/652240/13/gate/openstack-tox-py27/0504940/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/40/652240/13/gate/openstack-tox-py36/deef1fc/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/40/652240/13/gate/openstack-tox-py37/b4298ab/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/40/652240/13/gate/openstack-tox-docs/518e0cc/html/ : SUCCESS in 3m 26s\n- zun-tempest-docker-sql http://logs.openstack.org/40/652240/13/gate/zun-tempest-docker-sql/3c04136/ : SUCCESS in 46m 45s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/40/652240/13/gate/zun-tempest-py3-docker-sql/601b663/ : SUCCESS in 45m 16s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/40/652240/13/gate/zun-tempest-multinode-docker-sql/6695719/ : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":13},{"id":"b3fcbbc247f74af2cca33ffb4e201c98c84acac6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 03:30:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"8ce2a4f3a912d1223b351fe4ab7e8e9cb485386e","revisions":{"ab9aa445f11f689f2583044c3f9ddd66c824fbf7":{"kind":"REWORK","_number":1,"created":"2019-04-13 23:09:36.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/1"}}},"commit":{"parents":[{"commit":"9f98a16566192251892d58030a306bdb1e9f2a53","subject":"Merge \"Add validation for capsule template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9f98a16566192251892d58030a306bdb1e9f2a53"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"subject":"[WIP] Record action finish","message":"[WIP] Record action finish\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ab9aa445f11f689f2583044c3f9ddd66c824fbf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ab9aa445f11f689f2583044c3f9ddd66c824fbf7"}]},"branch":"refs/heads/master"},"aa8dcf6f3b4c558348b9e91a71a5c49a0214f1dc":{"kind":"REWORK","_number":2,"created":"2019-04-14 16:04:12.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/2"}}},"commit":{"parents":[{"commit":"9f98a16566192251892d58030a306bdb1e9f2a53","subject":"Merge \"Add validation for capsule template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9f98a16566192251892d58030a306bdb1e9f2a53"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-14 16:04:01.000000000","tz":0},"subject":"[WIP] Record action finish","message":"[WIP] Record action finish\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/aa8dcf6f3b4c558348b9e91a71a5c49a0214f1dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/aa8dcf6f3b4c558348b9e91a71a5c49a0214f1dc"}]},"branch":"refs/heads/master"},"1c32fb4816fbc5b04e338a10c14e8e7bcc99fd9a":{"kind":"REWORK","_number":3,"created":"2019-04-14 19:01:10.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/3"}}},"commit":{"parents":[{"commit":"9f98a16566192251892d58030a306bdb1e9f2a53","subject":"Merge \"Add validation for capsule template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9f98a16566192251892d58030a306bdb1e9f2a53"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-14 18:58:28.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1c32fb4816fbc5b04e338a10c14e8e7bcc99fd9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1c32fb4816fbc5b04e338a10c14e8e7bcc99fd9a"}]},"branch":"refs/heads/master"},"f5a80afada2696bd1601b9dd8dcad94eb5efe2aa":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-04-14 21:53:47.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/4"}}},"commit":{"parents":[{"commit":"9cbb95a4e4dce312f3bc4b9446888964b87c36ab","subject":"Use ISO8601 format for timestamp fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9cbb95a4e4dce312f3bc4b9446888964b87c36ab"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-14 21:50:58.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f5a80afada2696bd1601b9dd8dcad94eb5efe2aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f5a80afada2696bd1601b9dd8dcad94eb5efe2aa"}]},"branch":"refs/heads/master"},"14b96e8d01e65eca55550de83305ea9bfd41fbb8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-04-15 00:23:28.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/5"}}},"commit":{"parents":[{"commit":"203d35d1a8b8b58c76ade96643317d92d7afdae8","subject":"Use ISO8601 format for timestamp fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/203d35d1a8b8b58c76ade96643317d92d7afdae8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-15 00:18:44.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/14b96e8d01e65eca55550de83305ea9bfd41fbb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/14b96e8d01e65eca55550de83305ea9bfd41fbb8"}]},"branch":"refs/heads/master"},"853a2259cb779ac39e11602e832debf9b9033823":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-04-15 01:14:19.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/6"}}},"commit":{"parents":[{"commit":"fe6f9e7a4dbca1b1e6336e7ae402f4af65cd38e8","subject":"Use ISO8601 format for timestamp fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/fe6f9e7a4dbca1b1e6336e7ae402f4af65cd38e8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-15 01:14:02.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/853a2259cb779ac39e11602e832debf9b9033823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/853a2259cb779ac39e11602e832debf9b9033823"}]},"branch":"refs/heads/master"},"09fd5f7d3eaf9c8b6ff5b89dbff59a2fcab3d957":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-04-20 20:15:27.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/7"}}},"commit":{"parents":[{"commit":"b3d1949744a91c4ea3d01c3be602feb85cec2702","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b3d1949744a91c4ea3d01c3be602feb85cec2702"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-20 20:15:27.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/09fd5f7d3eaf9c8b6ff5b89dbff59a2fcab3d957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/09fd5f7d3eaf9c8b6ff5b89dbff59a2fcab3d957"}]},"branch":"refs/heads/master"},"3093e7510145c4044192f0cda1aa9a736301c8a1":{"kind":"REWORK","_number":8,"created":"2019-04-21 00:02:46.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/8"}}},"commit":{"parents":[{"commit":"b3d1949744a91c4ea3d01c3be602feb85cec2702","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b3d1949744a91c4ea3d01c3be602feb85cec2702"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 00:02:31.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3093e7510145c4044192f0cda1aa9a736301c8a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3093e7510145c4044192f0cda1aa9a736301c8a1"}]},"branch":"refs/heads/master"},"9a46b2613cea7520b99a3f09db4fd84de6961657":{"kind":"REWORK","_number":9,"created":"2019-04-21 02:53:35.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/9"}}},"commit":{"parents":[{"commit":"b3d1949744a91c4ea3d01c3be602feb85cec2702","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b3d1949744a91c4ea3d01c3be602feb85cec2702"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 02:53:27.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9a46b2613cea7520b99a3f09db4fd84de6961657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9a46b2613cea7520b99a3f09db4fd84de6961657"}]},"branch":"refs/heads/master"},"d5f7ca7d0a1632dbff1c4bd718dee1dba7b91fc4":{"kind":"REWORK","_number":10,"created":"2019-04-24 01:18:11.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/10"}}},"commit":{"parents":[{"commit":"56a589acc3e00880fb801dbf722d9d79ed3a688d","subject":"Trivial fix: rename instance_uuid to container_uuid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/56a589acc3e00880fb801dbf722d9d79ed3a688d"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-24 01:18:00.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d5f7ca7d0a1632dbff1c4bd718dee1dba7b91fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d5f7ca7d0a1632dbff1c4bd718dee1dba7b91fc4"}]},"branch":"refs/heads/master"},"5634ad86de150f379298cfb3635d216b0724d29a":{"kind":"REWORK","_number":11,"created":"2019-04-27 02:50:14.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/11"}}},"commit":{"parents":[{"commit":"4b8311aa1632263e7f6c2acfe0895d5ed3165e37","subject":"Merge \"Update task state in context manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4b8311aa1632263e7f6c2acfe0895d5ed3165e37"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-27 02:50:05.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5634ad86de150f379298cfb3635d216b0724d29a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5634ad86de150f379298cfb3635d216b0724d29a"}]},"branch":"refs/heads/master"},"3860c85c85192580877c4982f943cb90ae2aa27d":{"kind":"REWORK","_number":12,"created":"2019-05-18 16:02:50.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/12"}}},"commit":{"parents":[{"commit":"d1684a67950b6dcdaca013a0c1a920b76aaed3f8","subject":"Merge \"Remove etcd db backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1684a67950b6dcdaca013a0c1a920b76aaed3f8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-18 16:02:43.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3860c85c85192580877c4982f943cb90ae2aa27d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3860c85c85192580877c4982f943cb90ae2aa27d"}]},"branch":"refs/heads/master"},"8ce2a4f3a912d1223b351fe4ab7e8e9cb485386e":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-05-25 03:57:50.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/652240/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/40/652240/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/40/652240/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/40/652240/13"}}},"commit":{"parents":[{"commit":"f473532b8fd98f6eba74f7f1541445282af7a202","subject":"Merge \"api-ref: add missing parameters for container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/f473532b8fd98f6eba74f7f1541445282af7a202"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-13 23:09:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-25 03:57:50.000000000","tz":0},"subject":"Record action finish","message":"Record action finish\n\nIn before, there is no way to detect if an action is finished or not.\nThis commit add tracking for each action and record on DB once\nan action is finish.\n\nChange-Id: I9945087cf8d1484c09e60d11b1ba0c84c1799f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8ce2a4f3a912d1223b351fe4ab7e8e9cb485386e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8ce2a4f3a912d1223b351fe4ab7e8e9cb485386e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
