)]}'
{"id":"openstack%2Fzun~654071","triplet_id":"openstack%2Fzun~master~I79e0ac59d5306a2a646c545a24acc0be7e17e85c","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I79e0ac59d5306a2a646c545a24acc0be7e17e85c","subject":"Trivial fix: rename instance_uuid to container_uuid","status":"MERGED","created":"2019-04-21 00:16:20.000000000","updated":"2019-04-23 02:29:42.000000000","submitted":"2019-04-23 02:29:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654071-1555986582185-d1a6a7d1","meta_rev_id":"ffdf82335090254eb1677783fff43be7d98b2239","_number":654071,"virtual_id_number":654071,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-04-23 02:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":2,"date":"2019-04-23 01:17:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":1,"date":"2019-04-23 01:17:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 01:17:43.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-04-23 02:29:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66e4e62f422648e383e5fc3f16752617a5746fe4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-21 00:16:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e4b61f00723957f6a7b6c9fd589e2731e0e2edb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 01:23:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/71/654071/1/check/openstack-tox-cover/f781534/cover/ : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/654071/1/check/openstack-tox-lower-constraints/f76513d/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/71/654071/1/check/openstack-tox-pep8/3184f2a/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/71/654071/1/check/openstack-tox-py27/d6dde18/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/71/654071/1/check/openstack-tox-py36/a24a76a/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/71/654071/1/check/openstack-tox-py37/eb1cfe3/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/71/654071/1/check/openstack-tox-docs/f24b58c/html/ : SUCCESS in 3m 32s\n- zun-tempest-docker-sql http://logs.openstack.org/71/654071/1/check/zun-tempest-docker-sql/35c47aa/ : SUCCESS in 50m 19s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/71/654071/1/check/zun-tempest-py3-docker-sql/6f232b8/ : SUCCESS in 53m 30s\n- zun-tempest-docker-etcd http://logs.openstack.org/71/654071/1/check/zun-tempest-docker-etcd/efdcaa9/ : SUCCESS in 50m 08s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/71/654071/1/check/zun-tempest-multinode-docker-sql/2190e02/ : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4fd500e2b503f30b62cbed79a58ec31cb3481627","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-04-23 01:17:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1fde54072eb04877dcfdf8292b9bd6a27fbbbac6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:17:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"909310e19d2c4b8094e0d66f924447f6a4fdd9da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 02:29:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/71/654071/1/gate/openstack-tox-lower-constraints/b9cead9/ : SUCCESS in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/71/654071/1/gate/openstack-tox-pep8/86b232e/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/71/654071/1/gate/openstack-tox-py27/a9ed62d/ : SUCCESS in 4m 44s\n- openstack-tox-py36 http://logs.openstack.org/71/654071/1/gate/openstack-tox-py36/be7b4e1/ : SUCCESS in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/71/654071/1/gate/openstack-tox-py37/160887f/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/71/654071/1/gate/openstack-tox-docs/0364112/html/ : SUCCESS in 3m 14s\n- zun-tempest-docker-sql http://logs.openstack.org/71/654071/1/gate/zun-tempest-docker-sql/e72202f/ : SUCCESS in 41m 57s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/71/654071/1/gate/zun-tempest-py3-docker-sql/f9fcc51/ : SUCCESS in 49m 10s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/71/654071/1/gate/zun-tempest-multinode-docker-sql/d819d30/ : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ffdf82335090254eb1677783fff43be7d98b2239","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 02:29:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56a589acc3e00880fb801dbf722d9d79ed3a688d","revisions":{"56a589acc3e00880fb801dbf722d9d79ed3a688d":{"kind":"REWORK","_number":1,"created":"2019-04-21 00:16:20.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/71/654071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/71/654071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/654071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/654071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/71/654071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/71/654071/1"}}},"commit":{"parents":[{"commit":"b3d1949744a91c4ea3d01c3be602feb85cec2702","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b3d1949744a91c4ea3d01c3be602feb85cec2702"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 00:14:48.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 00:14:48.000000000","tz":0},"subject":"Trivial fix: rename instance_uuid to container_uuid","message":"Trivial fix: rename instance_uuid to container_uuid\n\nChange-Id: I79e0ac59d5306a2a646c545a24acc0be7e17e85c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/56a589acc3e00880fb801dbf722d9d79ed3a688d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/56a589acc3e00880fb801dbf722d9d79ed3a688d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
