)]}'
{"id":"openstack%2Fzun~654073","triplet_id":"openstack%2Fzun~master~I0fd83c684ffd2e7a7a5ac66017e2adda3c3ed113","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I0fd83c684ffd2e7a7a5ac66017e2adda3c3ed113","subject":"Allow overcommit of cpu resource in docker","status":"MERGED","created":"2019-04-21 02:49:19.000000000","updated":"2019-05-17 01:48:59.000000000","submitted":"2019-05-17 01:48:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654073-1558057739425-4b33e095","meta_rev_id":"b4d2c8cdc78ec8e016efc2a2ee12016ab014396c","_number":654073,"virtual_id_number":654073,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-05-17 01:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-05-17 00:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-05-17 00:44:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 00:44:22.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-05-17 01:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37da71c16505246966c93b13b5c13fc7f2c2abd9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-21 02:49:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f08b7ce64800dba75a1418db9278a592932524b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 04:46:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/73/654073/1/check/openstack-tox-cover/e3fc53b/ : FAILURE in 4m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/654073/1/check/openstack-tox-lower-constraints/2164f00/ : FAILURE in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/73/654073/1/check/openstack-tox-pep8/09f1c65/ : FAILURE in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/73/654073/1/check/openstack-tox-py27/275ef1e/ : FAILURE in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/73/654073/1/check/openstack-tox-py36/d6351e4/ : FAILURE in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/73/654073/1/check/openstack-tox-py37/f6a28f7/ : FAILURE in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/73/654073/1/check/openstack-tox-docs/a564991/html/ : SUCCESS in 3m 35s\n- zun-tempest-docker-sql http://logs.openstack.org/73/654073/1/check/zun-tempest-docker-sql/999cd01/ : TIMED_OUT in 1h 13m 28s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/73/654073/1/check/zun-tempest-py3-docker-sql/45d80cd/ : TIMED_OUT in 1h 13m 14s\n- zun-tempest-docker-etcd http://logs.openstack.org/73/654073/1/check/zun-tempest-docker-etcd/7328d85/ : TIMED_OUT in 1h 13m 17s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/73/654073/1/check/zun-tempest-multinode-docker-sql/b22f60d/ : FAILURE in 1h 52m 00s","accounts_in_message":[],"_revision_number":1},{"id":"44c206f8415b3dec1057ba20f734eea7c2074441","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-05 23:43:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"471dc45405815304228f10f7136077081ce2f802","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-05 23:50:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b4814cf47b2af76f68a649bd3deee0ee72a96e86","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-05 23:53:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"932d79367cabf911d9b559375ae3404ba30bddaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 01:00:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/73/654073/4/check/openstack-tox-cover/3df905b/cover/ : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/654073/4/check/openstack-tox-lower-constraints/4461b08/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/73/654073/4/check/openstack-tox-pep8/8057eef/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/73/654073/4/check/openstack-tox-py27/32e2792/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/73/654073/4/check/openstack-tox-py36/8233b6d/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/73/654073/4/check/openstack-tox-py37/15150ab/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/73/654073/4/check/openstack-tox-docs/745353f/html/ : SUCCESS in 3m 01s\n- zun-tempest-docker-sql http://logs.openstack.org/73/654073/4/check/zun-tempest-docker-sql/8c4b066/ : FAILURE in 51m 43s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/73/654073/4/check/zun-tempest-py3-docker-sql/6ef6ed3/ : FAILURE in 48m 32s\n- zun-tempest-docker-etcd http://logs.openstack.org/73/654073/4/check/zun-tempest-docker-etcd/228d0c8/ : FAILURE in 50m 18s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/73/654073/4/check/zun-tempest-multinode-docker-sql/d67a9da/ : FAILURE in 53m 58s","accounts_in_message":[],"_revision_number":4},{"id":"3c27c0cc4871acbbcfd3416807aa6358e39d1306","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-07 03:37:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"73fe770d8880a4ddb341d4b20509804444644f9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 04:43:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/73/654073/4/check/openstack-tox-cover/4f6a182/cover/ : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/654073/4/check/openstack-tox-lower-constraints/9389703/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/73/654073/4/check/openstack-tox-pep8/8b70354/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/73/654073/4/check/openstack-tox-py27/047d88f/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/73/654073/4/check/openstack-tox-py36/fdeb5a8/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/73/654073/4/check/openstack-tox-py37/1650da3/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/73/654073/4/check/openstack-tox-docs/66a7e0c/html/ : SUCCESS in 3m 33s\n- zun-tempest-docker-sql http://logs.openstack.org/73/654073/4/check/zun-tempest-docker-sql/08a18f2/ : FAILURE in 55m 14s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/73/654073/4/check/zun-tempest-py3-docker-sql/690c5f9/ : FAILURE in 43m 48s\n- zun-tempest-docker-etcd http://logs.openstack.org/73/654073/4/check/zun-tempest-docker-etcd/ea57cef/ : FAILURE in 41m 22s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/73/654073/4/check/zun-tempest-multinode-docker-sql/26c9ae6/ : FAILURE in 1h 04m 37s","accounts_in_message":[],"_revision_number":4},{"id":"0931ab6d3d662432f736aeba223ebd6a2b573af8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-11 17:58:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4ba1e2046fdcaa53eecb3fd0a419f9b59cdd79f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 19:03:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/73/654073/4/check/openstack-tox-cover/355aef4/cover/ : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/654073/4/check/openstack-tox-lower-constraints/ececafb/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/73/654073/4/check/openstack-tox-pep8/d1e9676/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/73/654073/4/check/openstack-tox-py27/c37c86a/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/73/654073/4/check/openstack-tox-py36/12368ec/ : SUCCESS in 3m 54s\n- openstack-tox-py37 http://logs.openstack.org/73/654073/4/check/openstack-tox-py37/1be6e6f/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/73/654073/4/check/openstack-tox-docs/f061380/html/ : SUCCESS in 3m 02s\n- zun-tempest-docker-sql http://logs.openstack.org/73/654073/4/check/zun-tempest-docker-sql/655cc83/ : SUCCESS in 49m 04s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/73/654073/4/check/zun-tempest-py3-docker-sql/61ac544/ : SUCCESS in 41m 34s\n- zun-tempest-docker-etcd http://logs.openstack.org/73/654073/4/check/zun-tempest-docker-etcd/6777a3e/ : SUCCESS in 48m 58s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/73/654073/4/check/zun-tempest-multinode-docker-sql/cf343bf/ : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":4},{"id":"639e9d3bd63bde283205c3e3d839936c6f0e612c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-11 22:24:51.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"24f2437797f7de5b2ebdd5a7501c171ad71033fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 23:18:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/73/654073/5/check/openstack-tox-cover/77a0580/cover/ : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/654073/5/check/openstack-tox-lower-constraints/3a3d34d/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/73/654073/5/check/openstack-tox-pep8/a3730e8/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/73/654073/5/check/openstack-tox-py27/a91f6ba/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/73/654073/5/check/openstack-tox-py36/85085dc/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/73/654073/5/check/openstack-tox-py37/9c83525/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/73/654073/5/check/openstack-tox-docs/cc76d18/html/ : SUCCESS in 3m 06s\n- zun-tempest-docker-sql http://logs.openstack.org/73/654073/5/check/zun-tempest-docker-sql/35692c0/ : SUCCESS in 47m 41s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/73/654073/5/check/zun-tempest-py3-docker-sql/e266bd4/ : SUCCESS in 47m 36s\n- zun-tempest-docker-etcd http://logs.openstack.org/73/654073/5/check/zun-tempest-docker-etcd/7994794/ : SUCCESS in 46m 37s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/73/654073/5/check/zun-tempest-multinode-docker-sql/93495ca/ : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":5},{"id":"8332a28e72dbb616507249ceba05bac18c7dfece","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-05-17 00:44:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d0563e36313ce6b2151245adc1677170c0f57df5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 00:44:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c85afd480a68b6a807ea85bc1dfaffb26cfad2de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 01:48:59.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/73/654073/5/gate/openstack-tox-lower-constraints/220f55d/ : SUCCESS in 5m 32s\n- openstack-tox-pep8 http://logs.openstack.org/73/654073/5/gate/openstack-tox-pep8/061e45b/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/73/654073/5/gate/openstack-tox-py27/a5e4733/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/73/654073/5/gate/openstack-tox-py36/3348ef7/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/73/654073/5/gate/openstack-tox-py37/653ec58/ : SUCCESS in 7m 04s\n- openstack-tox-docs http://logs.openstack.org/73/654073/5/gate/openstack-tox-docs/58b5e90/html/ : SUCCESS in 3m 20s\n- zun-tempest-docker-sql http://logs.openstack.org/73/654073/5/gate/zun-tempest-docker-sql/9243cd0/ : SUCCESS in 47m 47s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/73/654073/5/gate/zun-tempest-py3-docker-sql/3444ebc/ : SUCCESS in 49m 32s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/73/654073/5/gate/zun-tempest-multinode-docker-sql/677dc87/ : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":5},{"id":"b4d2c8cdc78ec8e016efc2a2ee12016ab014396c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 01:48:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0cb6140938612e95192e952339470d3fe0e67fa1","revisions":{"ee40612341e148f155c9de91f6ad3c9713c58957":{"kind":"REWORK","_number":1,"created":"2019-04-21 02:49:19.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/654073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/73/654073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/73/654073/1"}}},"commit":{"parents":[{"commit":"55067a04edab0a41d27b443034e8500c25ba5e90","subject":"Merge \"Use ISO8601 format for timestamp fields\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/55067a04edab0a41d27b443034e8500c25ba5e90"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 02:45:30.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 02:49:08.000000000","tz":0},"subject":"[WIP] Allow overcommit of cpu resource in docker","message":"[WIP] Allow overcommit of cpu resource in docker\n\nUse soft limit (the cpu share option) instead of hard limit (the\ncpu quota option) for CPU. This allows a container to use more\nCPU than its limits when the system is not full.\n\nChange-Id: I0fd83c684ffd2e7a7a5ac66017e2adda3c3ed113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ee40612341e148f155c9de91f6ad3c9713c58957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ee40612341e148f155c9de91f6ad3c9713c58957"}]},"branch":"refs/heads/master"},"ad38e490d864e584fb17a4f6ed94272633011c79":{"kind":"REWORK","_number":2,"created":"2019-05-05 23:43:09.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/654073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/73/654073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/73/654073/2"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 02:45:30.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 23:42:59.000000000","tz":0},"subject":"Allow overcommit of cpu resource in docker","message":"Allow overcommit of cpu resource in docker\n\nUse soft limit (the cpu share option) instead of hard limit (the\ncpu quota option) for CPU. This allows a container to use more\nCPU than its limits when the system is not full.\n\nChange-Id: I0fd83c684ffd2e7a7a5ac66017e2adda3c3ed113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ad38e490d864e584fb17a4f6ed94272633011c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/ad38e490d864e584fb17a4f6ed94272633011c79"}]},"branch":"refs/heads/master"},"c6fa9fa55b3eacc2234369f86bf8281521206266":{"kind":"REWORK","_number":3,"created":"2019-05-05 23:50:26.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/654073/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/73/654073/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/73/654073/3"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 02:45:30.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 23:50:21.000000000","tz":0},"subject":"Allow overcommit of cpu resource in docker","message":"Allow overcommit of cpu resource in docker\n\nUse soft limit (the cpu share option) instead of hard limit (the\ncpu quota option) for CPU. This allows a container to use more\nCPU than its limits when the system is not full.\n\nChange-Id: I0fd83c684ffd2e7a7a5ac66017e2adda3c3ed113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c6fa9fa55b3eacc2234369f86bf8281521206266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c6fa9fa55b3eacc2234369f86bf8281521206266"}]},"branch":"refs/heads/master"},"1af35972a6dc73b8f97b38ffe0cfd62bcfd8124c":{"kind":"REWORK","_number":4,"created":"2019-05-05 23:53:22.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/654073/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/73/654073/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/73/654073/4"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 02:45:30.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 23:53:15.000000000","tz":0},"subject":"Allow overcommit of cpu resource in docker","message":"Allow overcommit of cpu resource in docker\n\nUse soft limit (the cpu share option) instead of hard limit (the\ncpu quota option) for CPU. This allows a container to use more\nCPU than its limits when the system is not full.\n\nChange-Id: I0fd83c684ffd2e7a7a5ac66017e2adda3c3ed113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1af35972a6dc73b8f97b38ffe0cfd62bcfd8124c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1af35972a6dc73b8f97b38ffe0cfd62bcfd8124c"}]},"branch":"refs/heads/master"},"0cb6140938612e95192e952339470d3fe0e67fa1":{"kind":"NO_CHANGE","_number":5,"created":"2019-05-11 22:24:51.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/654073/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/73/654073/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/73/654073/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/73/654073/5"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 02:45:30.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-11 17:58:58.000000000","tz":0},"subject":"Allow overcommit of cpu resource in docker","message":"Allow overcommit of cpu resource in docker\n\nUse soft limit (the cpu share option) instead of hard limit (the\ncpu quota option) for CPU. This allows a container to use more\nCPU than its limits when the system is not full.\n\nChange-Id: I0fd83c684ffd2e7a7a5ac66017e2adda3c3ed113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0cb6140938612e95192e952339470d3fe0e67fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0cb6140938612e95192e952339470d3fe0e67fa1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
