)]}'
{"id":"openstack%2Fzun~654672","triplet_id":"openstack%2Fzun~master~Ia53c87f65171794c86a7a14288dbbe4985e66250","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ia53c87f65171794c86a7a14288dbbe4985e66250","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 02:23:37.000000000","updated":"2019-04-26 02:56:29.000000000","submitted":"2019-04-26 02:56:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654672-1556247389128-1b918c92","meta_rev_id":"b882fc6e36b771daab76df9f0ae7072352984dfb","_number":654672,"virtual_id_number":654672,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-04-26 02:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-04-26 01:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-04-26 01:43:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 06:43:06.000000000","updated_by":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"reviewer":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"state":"REVIEWER"},{"updated":"2019-04-26 01:43:05.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2019-04-26 02:56:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"690afaeac4b16af24fc653034a60de38c5a01a75","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 02:23:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac4108d65954ace13ae4057acf6e0e61b515831a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 03:43:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/72/654672/1/check/test-release-openstack/e9df23a/ : SUCCESS in 2m 51s\n- openstack-tox-cover http://logs.openstack.org/72/654672/1/check/openstack-tox-cover/e1deee7/cover/ : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/654672/1/check/openstack-tox-lower-constraints/2f60fc9/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/72/654672/1/check/openstack-tox-pep8/cec2ed0/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/72/654672/1/check/openstack-tox-py27/18d854f/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/72/654672/1/check/openstack-tox-py36/da00d05/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/72/654672/1/check/openstack-tox-py37/7a58117/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/72/654672/1/check/openstack-tox-docs/524b494/html/ : SUCCESS in 4m 04s\n- build-openstack-releasenotes http://logs.openstack.org/72/654672/1/check/build-openstack-releasenotes/7e233bb/html/ : SUCCESS in 4m 04s\n- zun-tempest-docker-sql http://logs.openstack.org/72/654672/1/check/zun-tempest-docker-sql/fe42dfb/ : SUCCESS in 50m 35s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/72/654672/1/check/zun-tempest-py3-docker-sql/7caaf46/ : SUCCESS in 49m 19s\n- zun-tempest-docker-etcd http://logs.openstack.org/72/654672/1/check/zun-tempest-docker-etcd/556ffdc/ : SUCCESS in 50m 18s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/72/654672/1/check/zun-tempest-multinode-docker-sql/85446b6/ : SUCCESS in 1h 06m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4f2f253939c84eaf7dc1ae2015575a00e52423a5","author":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"date":"2019-04-23 06:43:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f40c42bcea8c9dbc11940e515fe750fbbd3b3eca","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-25 02:55:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3c330967896e07eb709dd664859d0cd2bb95871d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 04:12:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/72/654672/2/check/test-release-openstack/93b0d7b/ : SUCCESS in 2m 25s\n- openstack-tox-cover http://logs.openstack.org/72/654672/2/check/openstack-tox-cover/db69654/cover/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/654672/2/check/openstack-tox-lower-constraints/88c8fbe/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/72/654672/2/check/openstack-tox-pep8/dc3c504/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/72/654672/2/check/openstack-tox-py27/6f4992d/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/72/654672/2/check/openstack-tox-py36/7a7621b/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/72/654672/2/check/openstack-tox-py37/b505b86/ : SUCCESS in 4m 52s\n- openstack-tox-docs http://logs.openstack.org/72/654672/2/check/openstack-tox-docs/5b515c6/html/ : SUCCESS in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/72/654672/2/check/build-openstack-releasenotes/45d0f65/html/ : SUCCESS in 2m 48s\n- zun-tempest-docker-sql http://logs.openstack.org/72/654672/2/check/zun-tempest-docker-sql/a08ae46/ : SUCCESS in 50m 24s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/72/654672/2/check/zun-tempest-py3-docker-sql/b8bf84d/ : SUCCESS in 51m 14s\n- zun-tempest-docker-etcd http://logs.openstack.org/72/654672/2/check/zun-tempest-docker-etcd/42d07c6/ : SUCCESS in 50m 29s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/72/654672/2/check/zun-tempest-multinode-docker-sql/9763e91/ : SUCCESS in 1h 03m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8c799f738467216153d5cb7e2e013ab5472a639a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-26 01:43:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a81227aa0b920233e515ebdbcd8245847e59b249","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 01:43:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d8be29b2b9ef6b04cc778b7f68efaaf9a21e980","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 02:56:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/72/654672/2/gate/test-release-openstack/7857f6c/ : SUCCESS in 2m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/654672/2/gate/openstack-tox-lower-constraints/4ef5a51/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/72/654672/2/gate/openstack-tox-pep8/b3a2204/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/72/654672/2/gate/openstack-tox-py27/3318848/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/72/654672/2/gate/openstack-tox-py36/7d6d663/ : SUCCESS in 4m 38s\n- openstack-tox-py37 http://logs.openstack.org/72/654672/2/gate/openstack-tox-py37/910f0ae/ : SUCCESS in 6m 15s\n- openstack-tox-docs http://logs.openstack.org/72/654672/2/gate/openstack-tox-docs/f8a6100/html/ : SUCCESS in 3m 44s\n- build-openstack-releasenotes http://logs.openstack.org/72/654672/2/gate/build-openstack-releasenotes/e54ac26/html/ : SUCCESS in 2m 37s\n- zun-tempest-docker-sql http://logs.openstack.org/72/654672/2/gate/zun-tempest-docker-sql/9269561/ : SUCCESS in 47m 04s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/72/654672/2/gate/zun-tempest-py3-docker-sql/fa58edb/ : SUCCESS in 45m 42s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/72/654672/2/gate/zun-tempest-multinode-docker-sql/970c699/ : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":2},{"id":"b882fc6e36b771daab76df9f0ae7072352984dfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 02:56:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3da61dd1f1e576f922e5f59ab3e2d5b5218801ea","revisions":{"a24d55f700bd4f4dc41549f494436ee913c5e95f":{"kind":"REWORK","_number":1,"created":"2019-04-23 02:23:37.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/72/654672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/72/654672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/72/654672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/72/654672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/72/654672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/72/654672/1"}}},"commit":{"parents":[{"commit":"b3d1949744a91c4ea3d01c3be602feb85cec2702","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b3d1949744a91c4ea3d01c3be602feb85cec2702"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 02:23:15.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 02:23:35.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia53c87f65171794c86a7a14288dbbe4985e66250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a24d55f700bd4f4dc41549f494436ee913c5e95f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a24d55f700bd4f4dc41549f494436ee913c5e95f"}]},"branch":"refs/heads/master"},"3da61dd1f1e576f922e5f59ab3e2d5b5218801ea":{"kind":"REWORK","_number":2,"created":"2019-04-25 02:55:45.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/72/654672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/72/654672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/72/654672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/72/654672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/72/654672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/72/654672/2"}}},"commit":{"parents":[{"commit":"b3d1949744a91c4ea3d01c3be602feb85cec2702","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b3d1949744a91c4ea3d01c3be602feb85cec2702"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 02:23:15.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-25 02:55:41.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia53c87f65171794c86a7a14288dbbe4985e66250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3da61dd1f1e576f922e5f59ab3e2d5b5218801ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/3da61dd1f1e576f922e5f59ab3e2d5b5218801ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
