)]}'
{"id":"openstack%2Fzun~655863","triplet_id":"openstack%2Fzun~stable%2Fstein~I52ab37564a9d551f03dd48b27c35b8f8360c2cbc","project":"openstack/zun","branch":"stable/stein","topic":"bug/1825725-stable/stein","hashtags":[],"change_id":"I52ab37564a9d551f03dd48b27c35b8f8360c2cbc","subject":"Fix a TypeError on _get_network_info","status":"MERGED","created":"2019-04-26 10:17:49.000000000","updated":"2019-04-27 01:24:26.000000000","submitted":"2019-04-27 01:24:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655863-1556328266249-1b918c92","meta_rev_id":"f69a9dee6b826503eab6f97ff3dcac373975ea4f","_number":655863,"virtual_id_number":655863,"owner":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-04-27 01:24:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-04-27 00:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-04-27 00:19:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-27 00:19:50.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2019-04-27 01:24:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b179421785b95552633401feb5132d157dde9f09","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-04-26 10:17:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ad305a6794f131ee34af8d0e6aab9ef33c3f3b9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 11:25:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/63/655863/1/check/openstack-tox-cover/fde2002/cover/ : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/655863/1/check/openstack-tox-lower-constraints/6596a91/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/63/655863/1/check/openstack-tox-pep8/d9d9383/ : SUCCESS in 6m 44s\n- openstack-tox-py27 http://logs.openstack.org/63/655863/1/check/openstack-tox-py27/03c8318/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/63/655863/1/check/openstack-tox-py35/c43ecc6/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/63/655863/1/check/openstack-tox-py36/6efa425/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/63/655863/1/check/openstack-tox-py37/b2b4298/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/63/655863/1/check/openstack-tox-docs/e9058ae/html/ : SUCCESS in 3m 32s\n- zun-tempest-docker-sql http://logs.openstack.org/63/655863/1/check/zun-tempest-docker-sql/87c574d/ : SUCCESS in 46m 19s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/63/655863/1/check/zun-tempest-py35-docker-sql/357d206/ : SUCCESS in 44m 53s\n- zun-tempest-docker-etcd http://logs.openstack.org/63/655863/1/check/zun-tempest-docker-etcd/7a2cda4/ : SUCCESS in 56m 10s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/63/655863/1/check/zun-tempest-multinode-docker-sql/0489e27/ : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":1},{"id":"acc967b1c461e56426c436b7e22d19a6413ba7e3","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-27 00:19:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8424c852c126800f731bc17a6d55a85a76dbe398","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 00:20:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6010e9bd657a0bb3cbb73b893534d966917443f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 01:24:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/63/655863/1/gate/openstack-tox-lower-constraints/70d0f69/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 http://logs.openstack.org/63/655863/1/gate/openstack-tox-pep8/9a4ead1/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/63/655863/1/gate/openstack-tox-py27/8e0f463/ : SUCCESS in 4m 11s\n- openstack-tox-py35 http://logs.openstack.org/63/655863/1/gate/openstack-tox-py35/c63324b/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/63/655863/1/gate/openstack-tox-py36/38b3754/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/63/655863/1/gate/openstack-tox-py37/2c15303/ : SUCCESS in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/63/655863/1/gate/openstack-tox-docs/0d83c61/html/ : SUCCESS in 3m 36s\n- zun-tempest-docker-sql http://logs.openstack.org/63/655863/1/gate/zun-tempest-docker-sql/02bd461/ : SUCCESS in 42m 41s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/63/655863/1/gate/zun-tempest-py35-docker-sql/786da6f/ : SUCCESS in 51m 02s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/63/655863/1/gate/zun-tempest-multinode-docker-sql/5799d38/ : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d5cf1f1079a3536c47f74d587b4d25ba31666b87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 01:24:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"383a94a3f4ae21d777db1033921a6cdc282a9f40","revisions":{"383a94a3f4ae21d777db1033921a6cdc282a9f40":{"kind":"REWORK","_number":1,"created":"2019-04-26 10:17:49.000000000","uploader":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"ref":"refs/changes/63/655863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/63/655863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/63/655863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/63/655863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/63/655863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/63/655863/1"}}},"commit":{"parents":[{"commit":"9a58ed6f13baaf60693e4f1ddfd110fcc79f536f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9a58ed6f13baaf60693e4f1ddfd110fcc79f536f"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-21 15:43:01.000000000","tz":0},"committer":{"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","date":"2019-04-26 10:17:49.000000000","tz":0},"subject":"Fix a TypeError on _get_network_info","message":"Fix a TypeError on _get_network_info\n\nCloses-Bug: #1825725\nChange-Id: I52ab37564a9d551f03dd48b27c35b8f8360c2cbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/383a94a3f4ae21d777db1033921a6cdc282a9f40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/383a94a3f4ae21d777db1033921a6cdc282a9f40"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
