)]}'
{"id":"openstack%2Fzun~656079","triplet_id":"openstack%2Fzun~master~Ib5b58d03064f6aab12f674e71d3ffff8eb318857","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ib5b58d03064f6aab12f674e71d3ffff8eb318857","subject":"Remove etcd devstack config","status":"MERGED","created":"2019-04-27 17:59:57.000000000","updated":"2019-05-17 02:10:46.000000000","submitted":"2019-05-17 02:10:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656079-1558059046119-38c90c1d","meta_rev_id":"ec277b59654e342e6191511a7263d93806b02d68","_number":656079,"virtual_id_number":656079,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-05-17 02:10:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":2,"date":"2019-05-17 01:03:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":1,"date":"2019-05-17 01:03:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 01:03:29.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-05-17 02:10:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fd5cfcda33cc5beec87c3d15316c175a185426b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-27 17:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2c865bc301af73c089d69572bf0b5506562796f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-27 18:11:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"96c8aa68a81bc9c8bc5f8eba4d1eb71dea5033ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 19:24:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/79/656079/2/check/openstack-tox-cover/05f519f/cover/ : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/656079/2/check/openstack-tox-lower-constraints/044ee91/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/79/656079/2/check/openstack-tox-pep8/13a3fe7/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/79/656079/2/check/openstack-tox-py27/8c08f4d/ : SUCCESS in 5m 31s\n- openstack-tox-py36 http://logs.openstack.org/79/656079/2/check/openstack-tox-py36/947bb98/ : SUCCESS in 5m 28s\n- openstack-tox-py37 http://logs.openstack.org/79/656079/2/check/openstack-tox-py37/9e071e7/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/79/656079/2/check/openstack-tox-docs/fa91ae3/html/ : SUCCESS in 4m 27s\n- zun-tempest-docker-sql http://logs.openstack.org/79/656079/2/check/zun-tempest-docker-sql/5b27e32/ : SUCCESS in 50m 43s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/79/656079/2/check/zun-tempest-py3-docker-sql/3632d25/ : SUCCESS in 49m 40s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/79/656079/2/check/zun-tempest-multinode-docker-sql/73c88e8/ : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":2},{"id":"f1a1e816d3fe2cfff4abfa07d4bd1600ebd3696a","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-05-17 01:03:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0028c4d441512a7485a9b6559c02dd5bdccc2e78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 01:03:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"587107543e2036bc21288141bf3b9df3d950ecd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:10:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/79/656079/2/gate/openstack-tox-lower-constraints/db0d439/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/79/656079/2/gate/openstack-tox-pep8/bad44a9/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/79/656079/2/gate/openstack-tox-py27/107bb26/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/79/656079/2/gate/openstack-tox-py36/4ae44f7/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/79/656079/2/gate/openstack-tox-py37/b45a670/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/79/656079/2/gate/openstack-tox-docs/f1debcf/html/ : SUCCESS in 4m 08s\n- zun-tempest-docker-sql http://logs.openstack.org/79/656079/2/gate/zun-tempest-docker-sql/66f3cab/ : SUCCESS in 47m 57s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/79/656079/2/gate/zun-tempest-py3-docker-sql/f0a8b8e/ : SUCCESS in 48m 37s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/79/656079/2/gate/zun-tempest-multinode-docker-sql/87fdcb4/ : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ec277b59654e342e6191511a7263d93806b02d68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:10:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"271ca60ce28050624431685c37b9ea31df03c594","revisions":{"684caf2fdc747eb34d253028dfef73f81499966b":{"kind":"REWORK","_number":1,"created":"2019-04-27 17:59:57.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/79/656079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/79/656079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/79/656079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/79/656079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/79/656079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/79/656079/1"}}},"commit":{"parents":[{"commit":"39f5dacc7df662e231e660a9ebdaaa9ea4e9cc05","subject":"Remove etcd job from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/39f5dacc7df662e231e660a9ebdaaa9ea4e9cc05"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-27 17:59:01.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-27 17:59:01.000000000","tz":0},"subject":"Remove etcd devstack config","message":"Remove etcd devstack config\n\nThe etcd backend is duplicated for removal. This commit removes\nthe etcd devstack related code.\n\nChange-Id: Ib5b58d03064f6aab12f674e71d3ffff8eb318857\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/684caf2fdc747eb34d253028dfef73f81499966b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/684caf2fdc747eb34d253028dfef73f81499966b"}]},"branch":"refs/heads/master"},"271ca60ce28050624431685c37b9ea31df03c594":{"kind":"REWORK","_number":2,"created":"2019-04-27 18:11:54.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/79/656079/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/79/656079/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/79/656079/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/79/656079/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/79/656079/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/79/656079/2"}}},"commit":{"parents":[{"commit":"39f5dacc7df662e231e660a9ebdaaa9ea4e9cc05","subject":"Remove etcd job from gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/39f5dacc7df662e231e660a9ebdaaa9ea4e9cc05"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-27 17:59:01.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-27 18:01:59.000000000","tz":0},"subject":"Remove etcd devstack config","message":"Remove etcd devstack config\n\nThe etcd backend is duplicated for removal. This commit removes\nthe etcd devstack related code.\n\nChange-Id: Ib5b58d03064f6aab12f674e71d3ffff8eb318857\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/271ca60ce28050624431685c37b9ea31df03c594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/271ca60ce28050624431685c37b9ea31df03c594"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
