)]}'
{"id":"openstack%2Fzun~656175","triplet_id":"openstack%2Fzun~master~I3a65f7bff5b342974c61fe853a254bec40d2d143","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I3a65f7bff5b342974c61fe853a254bec40d2d143","subject":"Add kolla-ansible Zun job to the gate","status":"MERGED","created":"2019-04-28 15:42:28.000000000","updated":"2019-08-22 02:42:11.000000000","submitted":"2019-05-17 02:01:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656175-1558058479947-4b33e095","meta_rev_id":"92a05acae51b7fddb5eee523030c2685fcb71443","_number":656175,"virtual_id_number":656175,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-05-17 02:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":2,"date":"2019-05-17 01:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":1,"date":"2019-05-17 01:01:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 01:01:55.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-05-17 02:01:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57eb740686ccc71e57da098f41096ab0182f1196","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-28 15:42:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c945b26f25b304fb0eee220e3bc1a724a9c65ed0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 16:43:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/75/656175/1/check/openstack-tox-cover/aec1c0c/cover/ : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/656175/1/check/openstack-tox-lower-constraints/a380df5/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/75/656175/1/check/openstack-tox-pep8/a4be0c0/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/75/656175/1/check/openstack-tox-py27/988a34d/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/75/656175/1/check/openstack-tox-py36/dcca8c7/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/75/656175/1/check/openstack-tox-py37/d32aa3d/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/75/656175/1/check/openstack-tox-docs/e769be7/html/ : SUCCESS in 3m 35s\n- zun-tempest-docker-sql http://logs.openstack.org/75/656175/1/check/zun-tempest-docker-sql/4e417cd/ : SUCCESS in 46m 23s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/75/656175/1/check/zun-tempest-py3-docker-sql/e72a7cb/ : SUCCESS in 46m 42s\n- zun-tempest-docker-etcd http://logs.openstack.org/75/656175/1/check/zun-tempest-docker-etcd/c133259/ : SUCCESS in 44m 13s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/75/656175/1/check/zun-tempest-multinode-docker-sql/6a7f0ac/ : SUCCESS in 59m 54s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/75/656175/1/check/kolla-ansible-ubuntu-source-zun/c1ebdfe/ : SUCCESS in 38m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"858f52fcb7381d4d83998ca01909294c968e3155","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-05-17 01:01:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3aa06174fc8e072322d9934743a07136077b096","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 01:02:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ca22be896108c47339bc55c85ab2fc5c0a08c44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:01:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/75/656175/1/gate/openstack-tox-lower-constraints/5d64b88/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/75/656175/1/gate/openstack-tox-pep8/edc0583/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/75/656175/1/gate/openstack-tox-py27/b598952/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/75/656175/1/gate/openstack-tox-py36/8b48353/ : SUCCESS in 5m 13s\n- openstack-tox-py37 http://logs.openstack.org/75/656175/1/gate/openstack-tox-py37/80e2987/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/75/656175/1/gate/openstack-tox-docs/733dc59/html/ : SUCCESS in 4m 25s\n- zun-tempest-docker-sql http://logs.openstack.org/75/656175/1/gate/zun-tempest-docker-sql/4cce704/ : SUCCESS in 47m 03s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/75/656175/1/gate/zun-tempest-py3-docker-sql/0b6ad6d/ : SUCCESS in 45m 59s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/75/656175/1/gate/zun-tempest-multinode-docker-sql/7c442f7/ : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":1},{"id":"361563feed8b375cad1b2fc57f43b06e086f0782","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:01:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"92a05acae51b7fddb5eee523030c2685fcb71443","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-22 02:42:11.000000000","message":"Created a revert of this change as Ib8a08cf969cb4eadef729c61acfed0ce925d947d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f93b06938be0930cf648560fd012b160f3a27a9","revisions":{"4f93b06938be0930cf648560fd012b160f3a27a9":{"kind":"REWORK","_number":1,"created":"2019-04-28 15:42:28.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/75/656175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/75/656175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/75/656175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/75/656175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/75/656175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/75/656175/1"}}},"commit":{"parents":[{"commit":"4b8311aa1632263e7f6c2acfe0895d5ed3165e37","subject":"Merge \"Update task state in context manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4b8311aa1632263e7f6c2acfe0895d5ed3165e37"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-28 15:41:27.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-28 15:41:27.000000000","tz":0},"subject":"Add kolla-ansible Zun job to the gate","message":"Add kolla-ansible Zun job to the gate\n\nChange-Id: I3a65f7bff5b342974c61fe853a254bec40d2d143\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4f93b06938be0930cf648560fd012b160f3a27a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4f93b06938be0930cf648560fd012b160f3a27a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
