)]}'
{"id":"openstack%2Fzun~657165","triplet_id":"openstack%2Fzun~master~I67ad3fd69bc7680fb565bfc138bc0d8b77157649","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I67ad3fd69bc7680fb565bfc138bc0d8b77157649","subject":"Disable image validation","status":"MERGED","created":"2019-05-04 21:01:28.000000000","updated":"2019-05-05 02:08:11.000000000","submitted":"2019-05-05 02:08:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657165-1557022091754-eabd120a","meta_rev_id":"44195300ea4f68343633faa8b83ed72c8712e7af","_number":657165,"virtual_id_number":657165,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-05-05 02:08:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-05-04 23:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-05-04 23:52:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-05 02:08:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b39ab847099da99feba31782a56644ab388b6e7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-04 21:01:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95611bead3c1aa29984f2d2db72113e5f520d4b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 22:15:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/65/657165/1/check/openstack-tox-cover/b43998c/ : FAILURE in 3m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/657165/1/check/openstack-tox-lower-constraints/62cd57e/ : FAILURE in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/65/657165/1/check/openstack-tox-pep8/869bef4/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/65/657165/1/check/openstack-tox-py27/a3b2853/ : FAILURE in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/65/657165/1/check/openstack-tox-py36/fc2dab3/ : FAILURE in 3m 52s\n- openstack-tox-py37 http://logs.openstack.org/65/657165/1/check/openstack-tox-py37/ac958c1/ : FAILURE in 6m 18s\n- openstack-tox-docs http://logs.openstack.org/65/657165/1/check/openstack-tox-docs/e5a093b/html/ : SUCCESS in 4m 17s\n- zun-tempest-docker-sql http://logs.openstack.org/65/657165/1/check/zun-tempest-docker-sql/4d0084e/ : SUCCESS in 1h 01m 11s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/65/657165/1/check/zun-tempest-py3-docker-sql/b878af6/ : SUCCESS in 48m 26s\n- zun-tempest-docker-etcd http://logs.openstack.org/65/657165/1/check/zun-tempest-docker-etcd/a34eac8/ : SUCCESS in 1h 07m 22s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/65/657165/1/check/zun-tempest-multinode-docker-sql/abcbba8/ : SUCCESS in 1h 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"82697692b5c167c07cffda70df0344650edeb874","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-04 22:38:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"227660e5918182578573ea7e6beee5a36bac48fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 23:36:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/65/657165/2/check/openstack-tox-cover/41e690a/cover/ : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/657165/2/check/openstack-tox-lower-constraints/adde43a/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/65/657165/2/check/openstack-tox-pep8/d1b5bdb/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/65/657165/2/check/openstack-tox-py27/025df38/ : SUCCESS in 4m 10s\n- openstack-tox-py36 http://logs.openstack.org/65/657165/2/check/openstack-tox-py36/d343001/ : SUCCESS in 4m 28s\n- openstack-tox-py37 http://logs.openstack.org/65/657165/2/check/openstack-tox-py37/8bf9f2b/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/65/657165/2/check/openstack-tox-docs/f8ccfc4/html/ : SUCCESS in 3m 42s\n- zun-tempest-docker-sql http://logs.openstack.org/65/657165/2/check/zun-tempest-docker-sql/d9f3df1/ : SUCCESS in 51m 03s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/65/657165/2/check/zun-tempest-py3-docker-sql/9bd499a/ : SUCCESS in 43m 00s\n- zun-tempest-docker-etcd http://logs.openstack.org/65/657165/2/check/zun-tempest-docker-etcd/77ca2c3/ : SUCCESS in 51m 40s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/65/657165/2/check/zun-tempest-multinode-docker-sql/c9ab19c/ : FAILURE in 34m 04s","accounts_in_message":[],"_revision_number":2},{"id":"fe1c7c7e6689d921d0dd0317ce9c0eb9c3b1782f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-04 23:50:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ce8459b12a998052b7abaa93670d14580b59e86b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-04 23:52:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThis unblock the gate. Fast approved.","accounts_in_message":[],"_revision_number":2},{"id":"339c00e88ad0529f62ff0d78e32192d6bc729e9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 01:01:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/65/657165/2/check/openstack-tox-cover/4a8f381/cover/ : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/657165/2/check/openstack-tox-lower-constraints/93a4917/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/65/657165/2/check/openstack-tox-pep8/ef2bf6e/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/65/657165/2/check/openstack-tox-py27/6d58162/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/65/657165/2/check/openstack-tox-py36/74228ed/ : SUCCESS in 4m 49s\n- openstack-tox-py37 http://logs.openstack.org/65/657165/2/check/openstack-tox-py37/85f2bbd/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/65/657165/2/check/openstack-tox-docs/75352ad/html/ : SUCCESS in 5m 18s\n- zun-tempest-docker-sql http://logs.openstack.org/65/657165/2/check/zun-tempest-docker-sql/92e001c/ : SUCCESS in 1h 01m 41s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/65/657165/2/check/zun-tempest-py3-docker-sql/e590dd2/ : SUCCESS in 48m 50s\n- zun-tempest-docker-etcd http://logs.openstack.org/65/657165/2/check/zun-tempest-docker-etcd/e99ceb6/ : SUCCESS in 1h 03m 05s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/65/657165/2/check/zun-tempest-multinode-docker-sql/d50cd7e/ : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":2},{"id":"97d7c823d47ed3c743ae67eea0128b7b8c869697","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 01:01:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b2feca6a54c7e0293d252ca34bdb2f47fbb5db63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 02:08:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/657165/2/gate/openstack-tox-lower-constraints/8277d3c/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/65/657165/2/gate/openstack-tox-pep8/367cb8c/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/65/657165/2/gate/openstack-tox-py27/cd0ec91/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/65/657165/2/gate/openstack-tox-py36/afb1509/ : SUCCESS in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/65/657165/2/gate/openstack-tox-py37/2ebca9b/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/65/657165/2/gate/openstack-tox-docs/617a9cd/html/ : SUCCESS in 3m 23s\n- zun-tempest-docker-sql http://logs.openstack.org/65/657165/2/gate/zun-tempest-docker-sql/3378000/ : SUCCESS in 51m 30s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/65/657165/2/gate/zun-tempest-py3-docker-sql/cbf4558/ : SUCCESS in 46m 50s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/65/657165/2/gate/zun-tempest-multinode-docker-sql/9aee519/ : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":2},{"id":"44195300ea4f68343633faa8b83ed72c8712e7af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 02:08:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","revisions":{"8169585ebc6bb0557ec89993a7b6982bd38ebc51":{"kind":"REWORK","_number":1,"created":"2019-05-04 21:01:28.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/65/657165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/65/657165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/65/657165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/65/657165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/65/657165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/65/657165/1"}}},"commit":{"parents":[{"commit":"b3d1949744a91c4ea3d01c3be602feb85cec2702","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b3d1949744a91c4ea3d01c3be602feb85cec2702"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-04 20:57:25.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-04 20:57:25.000000000","tz":0},"subject":"Disable image validation","message":"Disable image validation\n\nIn before, Zun validate the container\u0027s image before lauching\nthe container. The goal is to have a pre-flight check before\ndoing an expensive operation.\n\nHowever, this feature has several disadvantages:\n* It blocks the zun api server on validating the image\n* It relies on API call to external image registry (i.e. DockerHub),\n  so it is not reliable.\n\nThis commit changes the default configuration to disable image\nvalidation by default.\n\nChange-Id: I67ad3fd69bc7680fb565bfc138bc0d8b77157649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8169585ebc6bb0557ec89993a7b6982bd38ebc51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8169585ebc6bb0557ec89993a7b6982bd38ebc51"}]},"branch":"refs/heads/master"},"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b":{"kind":"REWORK","_number":2,"created":"2019-05-04 22:38:47.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/65/657165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/65/657165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/65/657165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/65/657165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/65/657165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/65/657165/2"}}},"commit":{"parents":[{"commit":"4b8311aa1632263e7f6c2acfe0895d5ed3165e37","subject":"Merge \"Update task state in context manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4b8311aa1632263e7f6c2acfe0895d5ed3165e37"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-04 20:57:25.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-04 22:38:39.000000000","tz":0},"subject":"Disable image validation","message":"Disable image validation\n\nIn before, Zun validate the container\u0027s image before lauching\nthe container. The goal is to have a pre-flight check before\ndoing an expensive operation.\n\nHowever, this feature has several disadvantages:\n* It blocks the zun api server on validating the image\n* It relies on API call to external image registry (i.e. DockerHub),\n  so it is not reliable.\n\nThis commit changes the default configuration to disable image\nvalidation by default.\n\nChange-Id: I67ad3fd69bc7680fb565bfc138bc0d8b77157649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
