)]}'
{"id":"openstack%2Fzun~657285","triplet_id":"openstack%2Fzun~master~I901352aa39d40fafe039c96ece4ea45c92f3ffe8","project":"openstack/zun","branch":"master","topic":"portmapping","hashtags":[],"change_id":"I901352aa39d40fafe039c96ece4ea45c92f3ffe8","subject":"Support exposing ports in capsule - driver","status":"MERGED","created":"2019-05-05 20:37:36.000000000","updated":"2019-06-17 07:54:13.000000000","submitted":"2019-06-17 07:54:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657285-1560758053420-018dc456","meta_rev_id":"5fbcf683a41d38940a65064d581d8c12af503577","_number":657285,"virtual_id_number":657285,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":21428,"name":"JiWei","email":"17366060309@163.com","username":"WeiJi"},{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-06-17 07:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21428,"name":"JiWei","email":"17366060309@163.com","username":"WeiJi"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-06-17 06:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21428,"name":"JiWei","email":"17366060309@163.com","username":"WeiJi"},{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21428,"name":"JiWei","email":"17366060309@163.com","username":"WeiJi"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-06-17 06:48:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21428,"name":"JiWei","email":"17366060309@163.com","username":"WeiJi"},{"value":0,"date":"2019-06-08 14:54:22.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":21428,"name":"JiWei","email":"17366060309@163.com","username":"WeiJi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-08 14:54:22.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-06-17 06:48:37.000000000","updated_by":{"_account_id":21428,"name":"JiWei","email":"17366060309@163.com","username":"WeiJi"},"reviewer":{"_account_id":21428,"name":"JiWei","email":"17366060309@163.com","username":"WeiJi"},"state":"REVIEWER"},{"updated":"2019-06-17 07:54:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ff3a2b5f9d69d2f3c7fd29419c4e2cc1bbf246f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-05 20:37:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c42233ffb88247c3bd2a24e6f523a1273d56f2cb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-05 20:45:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2160c213c4afdfd5d9f5292e6f1e47a86819eb04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 22:32:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/85/657285/2/check/openstack-tox-cover/40924d6/cover/ : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/657285/2/check/openstack-tox-lower-constraints/5f64419/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/85/657285/2/check/openstack-tox-pep8/b814b7d/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/85/657285/2/check/openstack-tox-py27/fc5a579/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/85/657285/2/check/openstack-tox-py36/82a486b/ : SUCCESS in 5m 53s\n- openstack-tox-py37 http://logs.openstack.org/85/657285/2/check/openstack-tox-py37/bdbc664/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/85/657285/2/check/openstack-tox-docs/c7cba5e/html/ : SUCCESS in 3m 55s\n- zun-tempest-docker-sql http://logs.openstack.org/85/657285/2/check/zun-tempest-docker-sql/3c79b10/ : TIMED_OUT in 1h 12m 26s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/85/657285/2/check/zun-tempest-py3-docker-sql/526269e/ : TIMED_OUT in 1h 12m 11s\n- zun-tempest-docker-etcd http://logs.openstack.org/85/657285/2/check/zun-tempest-docker-etcd/de1618e/ : TIMED_OUT in 1h 12m 46s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/85/657285/2/check/zun-tempest-multinode-docker-sql/13b3d48/ : FAILURE in 1h 46m 07s","accounts_in_message":[],"_revision_number":2},{"id":"8548aa16cb17bda74090762c92a2cf3a243ec0ac","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-05 23:08:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"41f01099aa6b63fdffa02e839bcb02786d8c61ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 00:24:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/85/657285/3/check/openstack-tox-cover/164b8b2/cover/ : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/657285/3/check/openstack-tox-lower-constraints/ee1cc98/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/85/657285/3/check/openstack-tox-pep8/7f8dc7a/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/85/657285/3/check/openstack-tox-py27/1acbfbc/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/85/657285/3/check/openstack-tox-py36/eaaded3/ : SUCCESS in 5m 36s\n- openstack-tox-py37 http://logs.openstack.org/85/657285/3/check/openstack-tox-py37/e003ac6/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/85/657285/3/check/openstack-tox-docs/a4466b3/html/ : SUCCESS in 4m 14s\n- zun-tempest-docker-sql http://logs.openstack.org/85/657285/3/check/zun-tempest-docker-sql/c2efbab/ : SUCCESS in 50m 16s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/85/657285/3/check/zun-tempest-py3-docker-sql/8af247e/ : SUCCESS in 49m 42s\n- zun-tempest-docker-etcd http://logs.openstack.org/85/657285/3/check/zun-tempest-docker-etcd/3d1d2cd/ : SUCCESS in 50m 24s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/85/657285/3/check/zun-tempest-multinode-docker-sql/b778578/ : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":3},{"id":"630e515849bf5fd7d90fb917e3e1e9a428469000","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 16:05:02.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"586c1c42083bae8e24c327258bab31b617a63a10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 16:57:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/85/657285/4/check/openstack-tox-cover/7e70d3b/cover/ : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/657285/4/check/openstack-tox-lower-constraints/44bfa04/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/85/657285/4/check/openstack-tox-pep8/1a2dcf7/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/85/657285/4/check/openstack-tox-py27/9171505/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/85/657285/4/check/openstack-tox-py36/7631458/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/85/657285/4/check/openstack-tox-py37/677cafa/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/85/657285/4/check/openstack-tox-docs/b69f82b/html/ : SUCCESS in 3m 29s\n- zun-tempest-docker-sql http://logs.openstack.org/85/657285/4/check/zun-tempest-docker-sql/09cdfa2/ : SUCCESS in 46m 42s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/85/657285/4/check/zun-tempest-py3-docker-sql/1cb92c4/ : SUCCESS in 44m 13s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/85/657285/4/check/zun-tempest-multinode-docker-sql/056432f/ : SUCCESS in 51m 04s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/85/657285/4/check/kolla-ansible-ubuntu-source-zun/51b0a3f/ : SUCCESS in 41m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0b51ff4e79d5c93043a989b0834243bfcaece885","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-25 03:58:21.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"126eefae0b14b4f4e14142de8e983fddacf3baa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 04:55:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/85/657285/5/check/openstack-tox-cover/eac50ce/cover/ : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/657285/5/check/openstack-tox-lower-constraints/d07a674/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/85/657285/5/check/openstack-tox-pep8/c714fb8/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/85/657285/5/check/openstack-tox-py27/1f52a67/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/85/657285/5/check/openstack-tox-py36/d0a6ca3/ : SUCCESS in 3m 43s\n- openstack-tox-py37 http://logs.openstack.org/85/657285/5/check/openstack-tox-py37/18bc303/ : SUCCESS in 4m 21s\n- openstack-tox-docs http://logs.openstack.org/85/657285/5/check/openstack-tox-docs/e589bfd/html/ : SUCCESS in 3m 33s\n- zun-tempest-docker-sql http://logs.openstack.org/85/657285/5/check/zun-tempest-docker-sql/a38afab/ : SUCCESS in 51m 42s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/85/657285/5/check/zun-tempest-py3-docker-sql/20db39c/ : SUCCESS in 45m 28s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/85/657285/5/check/zun-tempest-multinode-docker-sql/255d45d/ : FAILURE in 56m 00s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/85/657285/5/check/kolla-ansible-ubuntu-source-zun/25bb2a5/ : SUCCESS in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"41c869b9a8911ca61e3742a5267c47e4e3ddaf4b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-25 15:47:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"456e559e5f6c3c9d2837d998f870ed16270e0853","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 16:55:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/85/657285/5/check/openstack-tox-cover/aeb5e19/cover/ : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/657285/5/check/openstack-tox-lower-constraints/b5b68a4/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/85/657285/5/check/openstack-tox-pep8/ddfa692/ : SUCCESS in 6m 23s\n- openstack-tox-py27 http://logs.openstack.org/85/657285/5/check/openstack-tox-py27/da8633a/ : SUCCESS in 6m 04s\n- openstack-tox-py36 http://logs.openstack.org/85/657285/5/check/openstack-tox-py36/d07d125/ : SUCCESS in 5m 09s\n- openstack-tox-py37 http://logs.openstack.org/85/657285/5/check/openstack-tox-py37/052afee/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/85/657285/5/check/openstack-tox-docs/a2048dc/html/ : SUCCESS in 4m 47s\n- zun-tempest-docker-sql http://logs.openstack.org/85/657285/5/check/zun-tempest-docker-sql/3449b43/ : SUCCESS in 52m 22s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/85/657285/5/check/zun-tempest-py3-docker-sql/2e8e83f/ : SUCCESS in 48m 35s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/85/657285/5/check/zun-tempest-multinode-docker-sql/f1aca9e/ : SUCCESS in 1h 02m 01s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/85/657285/5/check/kolla-ansible-ubuntu-source-zun/1983d51/ : SUCCESS in 39m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8e1351c2a9369253d410c14c2250da8608513100","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-01 14:54:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"35766d676088f5497e3edd2a1eaa41599ddb413d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 16:08:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/85/657285/6/check/openstack-tox-cover/195954d/cover/ : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/657285/6/check/openstack-tox-lower-constraints/828f645/ : SUCCESS in 6m 35s\n- openstack-tox-pep8 http://logs.openstack.org/85/657285/6/check/openstack-tox-pep8/7fd594d/ : SUCCESS in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/85/657285/6/check/openstack-tox-py27/5d42d5b/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/85/657285/6/check/openstack-tox-py36/1e0d88b/ : SUCCESS in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/85/657285/6/check/openstack-tox-py37/536760f/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/85/657285/6/check/openstack-tox-docs/8edfa17/html/ : SUCCESS in 3m 20s\n- zun-tempest-docker-sql http://logs.openstack.org/85/657285/6/check/zun-tempest-docker-sql/7e1e891/ : SUCCESS in 44m 10s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/85/657285/6/check/zun-tempest-py3-docker-sql/f1b8be2/ : SUCCESS in 46m 05s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/85/657285/6/check/zun-tempest-multinode-docker-sql/ea4466f/ : SUCCESS in 53m 09s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/85/657285/6/check/kolla-ansible-ubuntu-source-zun/1632a6b/ : SUCCESS in 40m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"80038c6d2699ee38b6083f8a428fbd71c7c65fc9","author":{"_account_id":21428,"name":"JiWei","email":"17366060309@163.com","username":"WeiJi"},"date":"2019-06-17 06:48:37.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"58b0fcd4a8f178207ace8db76bd998ac8958fec9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 06:48:49.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ac9ae85015d7fe0cac2f4b79dbd3f4b6219d760b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 07:54:13.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/657285/6/gate/openstack-tox-lower-constraints/717a795/ : SUCCESS in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/85/657285/6/gate/openstack-tox-pep8/7793a2a/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/85/657285/6/gate/openstack-tox-py27/466afda/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/85/657285/6/gate/openstack-tox-py36/c952de1/ : SUCCESS in 5m 51s\n- openstack-tox-py37 http://logs.openstack.org/85/657285/6/gate/openstack-tox-py37/3b67ef6/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/85/657285/6/gate/openstack-tox-docs/fed4e8b/html/ : SUCCESS in 3m 46s\n- zun-tempest-docker-sql http://logs.openstack.org/85/657285/6/gate/zun-tempest-docker-sql/8d82289/ : SUCCESS in 46m 52s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/85/657285/6/gate/zun-tempest-py3-docker-sql/a51c410/ : SUCCESS in 48m 14s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/85/657285/6/gate/zun-tempest-multinode-docker-sql/ecc2245/ : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":6},{"id":"02dd1f177cb43ac2f4fbece3bea4e258cbcf975b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 07:54:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"20ad7bbe0cc2f6a10ecf2d8ccff6888951a4da66","revisions":{"416028f6446390d86901c1db9b840697fbe73153":{"kind":"REWORK","_number":1,"created":"2019-05-05 20:37:36.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/85/657285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/85/657285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/85/657285/1"}}},"commit":{"parents":[{"commit":"4b8311aa1632263e7f6c2acfe0895d5ed3165e37","subject":"Merge \"Update task state in context manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4b8311aa1632263e7f6c2acfe0895d5ed3165e37"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 20:33:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 20:33:38.000000000","tz":0},"subject":"Support exposing ports in capsule - driver","message":"Support exposing ports in capsule - driver\n\nThis is a driver side implementation of port mapping for capsule.\nProcess the exposed_ports on creating each container in capsule.\n\nChange-Id: I901352aa39d40fafe039c96ece4ea45c92f3ffe8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/416028f6446390d86901c1db9b840697fbe73153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/416028f6446390d86901c1db9b840697fbe73153"}]},"branch":"refs/heads/master"},"c61cd99d6f1d711cc86f8d38d3740875af6c1e20":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-05 20:45:30.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/85/657285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/85/657285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/85/657285/2"}}},"commit":{"parents":[{"commit":"d42923f8cf6ea4a28cf11938b84455f34fa2b863","subject":"Add \u0027init_containers\u0027 field to capsule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d42923f8cf6ea4a28cf11938b84455f34fa2b863"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 20:33:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 20:45:16.000000000","tz":0},"subject":"Support exposing ports in capsule - driver","message":"Support exposing ports in capsule - driver\n\nThis is a driver side implementation of port mapping for capsule.\nProcess the exposed_ports on creating each container in capsule.\n\nChange-Id: I901352aa39d40fafe039c96ece4ea45c92f3ffe8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c61cd99d6f1d711cc86f8d38d3740875af6c1e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c61cd99d6f1d711cc86f8d38d3740875af6c1e20"}]},"branch":"refs/heads/master"},"1189e34a69c0d52d74e5fe69eb5235d8d6619fe8":{"kind":"REWORK","_number":3,"created":"2019-05-05 23:08:30.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/85/657285/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/85/657285/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/85/657285/3"}}},"commit":{"parents":[{"commit":"d42923f8cf6ea4a28cf11938b84455f34fa2b863","subject":"Add \u0027init_containers\u0027 field to capsule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d42923f8cf6ea4a28cf11938b84455f34fa2b863"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 20:33:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 23:05:09.000000000","tz":0},"subject":"Support exposing ports in capsule - driver","message":"Support exposing ports in capsule - driver\n\nThis is a driver side implementation of port mapping for capsule.\nProcess the exposed_ports on creating each container in capsule.\n\nChange-Id: I901352aa39d40fafe039c96ece4ea45c92f3ffe8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1189e34a69c0d52d74e5fe69eb5235d8d6619fe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/1189e34a69c0d52d74e5fe69eb5235d8d6619fe8"}]},"branch":"refs/heads/master"},"cb347661af74d0db10cb074e3c349c2897beb8b0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-18 16:05:02.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/85/657285/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/85/657285/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/85/657285/4"}}},"commit":{"parents":[{"commit":"b23e8cd25821651b8cde14bc6a9b15cd43a7fd49","subject":"Add \u0027init_containers\u0027 field to capsule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b23e8cd25821651b8cde14bc6a9b15cd43a7fd49"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 20:33:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-18 16:04:56.000000000","tz":0},"subject":"Support exposing ports in capsule - driver","message":"Support exposing ports in capsule - driver\n\nThis is a driver side implementation of port mapping for capsule.\nProcess the exposed_ports on creating each container in capsule.\n\nChange-Id: I901352aa39d40fafe039c96ece4ea45c92f3ffe8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/cb347661af74d0db10cb074e3c349c2897beb8b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/cb347661af74d0db10cb074e3c349c2897beb8b0"}]},"branch":"refs/heads/master"},"b3b690010a6753514ca47089bd26eb0739274c0a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-05-25 03:58:21.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/85/657285/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/85/657285/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/85/657285/5"}}},"commit":{"parents":[{"commit":"eed979185f14a86dcd21d72e2eb28bb9923619b7","subject":"Add \u0027init_containers\u0027 field to capsule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/eed979185f14a86dcd21d72e2eb28bb9923619b7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 20:33:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-25 03:58:21.000000000","tz":0},"subject":"Support exposing ports in capsule - driver","message":"Support exposing ports in capsule - driver\n\nThis is a driver side implementation of port mapping for capsule.\nProcess the exposed_ports on creating each container in capsule.\n\nChange-Id: I901352aa39d40fafe039c96ece4ea45c92f3ffe8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b3b690010a6753514ca47089bd26eb0739274c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b3b690010a6753514ca47089bd26eb0739274c0a"}]},"branch":"refs/heads/master"},"20ad7bbe0cc2f6a10ecf2d8ccff6888951a4da66":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-01 14:54:47.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/85/657285/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/85/657285/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/85/657285/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/85/657285/6"}}},"commit":{"parents":[{"commit":"83eaf006d6f9749ac85ce14de3a648be6d03011f","subject":"Merge \"Add \u0027init_containers\u0027 field to capsule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/83eaf006d6f9749ac85ce14de3a648be6d03011f"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 20:33:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-01 14:54:47.000000000","tz":0},"subject":"Support exposing ports in capsule - driver","message":"Support exposing ports in capsule - driver\n\nThis is a driver side implementation of port mapping for capsule.\nProcess the exposed_ports on creating each container in capsule.\n\nChange-Id: I901352aa39d40fafe039c96ece4ea45c92f3ffe8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/20ad7bbe0cc2f6a10ecf2d8ccff6888951a4da66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/20ad7bbe0cc2f6a10ecf2d8ccff6888951a4da66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
