)]}'
{"id":"openstack%2Fzun~657287","triplet_id":"openstack%2Fzun~master~I54bf68a6d9f3d79f47fbcefa55509bc0993027bb","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I54bf68a6d9f3d79f47fbcefa55509bc0993027bb","subject":"Trivial: fix a format issue in api history docs","status":"MERGED","created":"2019-05-05 22:55:25.000000000","updated":"2019-05-18 15:55:05.000000000","submitted":"2019-05-13 07:57:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657287-1557734272586-1a85bf1c","meta_rev_id":"a7a2ee809538adeace6920fc4c0a39ac069196b7","_number":657287,"virtual_id_number":657287,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-05-13 07:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":2,"date":"2019-05-13 07:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":1,"date":"2019-05-13 07:14:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 07:14:19.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-05-13 07:57:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52e304d148df9d5dca42a89f1a31bda8596bd484","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-05 22:55:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a12c2fa9dd288cdea136f8c4b7a39d6909c9a3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 22:59:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/657287/1/check/openstack-tox-pep8/df06187/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/87/657287/1/check/openstack-tox-docs/b7dea77/html/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"495fe2c0c39e3326e409ebe27467b46bc18fe357","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-05-13 07:14:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"edf88d9a4b43456009b8a0aea4432cf6301d0792","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 07:14:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ae59159981539d951c2732f71119280e10a7f83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 07:57:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/657287/1/gate/openstack-tox-pep8/a94932d/ : SUCCESS in 4m 13s\n- openstack-tox-docs http://logs.openstack.org/87/657287/1/gate/openstack-tox-docs/b120ac3/html/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fe838658992d88f10c8a37664c8c67284c1ecd17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 07:57:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7a2ee809538adeace6920fc4c0a39ac069196b7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 15:55:05.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 9bfbbd9a4b09c808305191de630ee531f3d8ca9f","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d62f9438ce7a1edaa29cdf22d360adfafed5c230","revisions":{"d62f9438ce7a1edaa29cdf22d360adfafed5c230":{"kind":"REWORK","_number":1,"created":"2019-05-05 22:55:25.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/87/657287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/87/657287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/87/657287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/87/657287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/87/657287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/87/657287/1"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 22:54:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 22:54:45.000000000","tz":0},"subject":"Trivial: fix a format issue in api history docs","message":"Trivial: fix a format issue in api history docs\n\nChange-Id: I54bf68a6d9f3d79f47fbcefa55509bc0993027bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d62f9438ce7a1edaa29cdf22d360adfafed5c230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d62f9438ce7a1edaa29cdf22d360adfafed5c230"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
