)]}'
{"id":"openstack%2Fzun~657517","triplet_id":"openstack%2Fzun~stable%2Frocky~I3c65cdef17d8279233bfbb74a38c9db79b01a2de","project":"openstack/zun","branch":"stable/rocky","topic":"bug/1827968","hashtags":[],"change_id":"I3c65cdef17d8279233bfbb74a38c9db79b01a2de","subject":"Skip image validation if failed","status":"MERGED","created":"2019-05-07 03:08:58.000000000","updated":"2019-05-11 15:48:12.000000000","submitted":"2019-05-11 15:48:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657517-1557589692479-6085ff31","meta_rev_id":"2228e8f8ec500237a01fafaea0b40285fc75ecf6","_number":657517,"virtual_id_number":657517,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-05-11 15:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-05-11 14:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-05-11 14:59:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-11 15:48:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72e024df3f89d081206b12e6d395d34b36aaf11b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-07 03:08:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac9c0742b10983ab3ec4e5a189266528e4edd6f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 04:48:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/17/657517/1/check/openstack-tox-pep8/654e957/ : SUCCESS in 8m 33s\n- openstack-tox-py27 http://logs.openstack.org/17/657517/1/check/openstack-tox-py27/e9078b7/ : SUCCESS in 6m 31s\n- openstack-tox-py35 http://logs.openstack.org/17/657517/1/check/openstack-tox-py35/878156a/ : SUCCESS in 6m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/17/657517/1/check/build-openstack-sphinx-docs/eb2c2f6/html/ : SUCCESS in 6m 22s\n- zun-tempest-docker-sql http://logs.openstack.org/17/657517/1/check/zun-tempest-docker-sql/0b6d4b3/ : SUCCESS in 53m 49s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/17/657517/1/check/zun-tempest-py35-docker-sql/005de99/ : SUCCESS in 54m 50s\n- zun-tempest-docker-etcd http://logs.openstack.org/17/657517/1/check/zun-tempest-docker-etcd/a4412c8/ : SUCCESS in 54m 50s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/17/657517/1/check/zun-tempest-multinode-docker-sql/f9a61a4/ : FAILURE in 1h 28m 04s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/17/657517/1/check/openstack-tox-lower-constraints/e440a47/ : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"378d77abc016ccf67c2c91c53f568f09f9ff8cce","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-11 14:59:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e6d27f4c40ec7d57107f945b2ef172e8b4c684d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 14:59:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a34c567209d499273536964d4a6974736241e8a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 15:48:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"985988c7244690c9895ae2ce1286f0734a6fa8c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 15:48:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/17/657517/1/gate/openstack-tox-pep8/a6261ea/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/17/657517/1/gate/openstack-tox-py27/caf6fa2/ : SUCCESS in 5m 10s\n- openstack-tox-py35 http://logs.openstack.org/17/657517/1/gate/openstack-tox-py35/3c56f56/ : SUCCESS in 5m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/17/657517/1/gate/build-openstack-sphinx-docs/71bb488/html/ : SUCCESS in 4m 54s\n- zun-tempest-docker-sql http://logs.openstack.org/17/657517/1/gate/zun-tempest-docker-sql/13ac06b/ : SUCCESS in 47m 06s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/17/657517/1/gate/zun-tempest-py35-docker-sql/a9392b9/ : SUCCESS in 44m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/657517/1/gate/openstack-tox-lower-constraints/8f63335/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b9c0581cce2e0b9d2272fa734c744e6dbea3d3e","revisions":{"2b9c0581cce2e0b9d2272fa734c744e6dbea3d3e":{"kind":"REWORK","_number":1,"created":"2019-05-07 03:08:58.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/17/657517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/17/657517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/17/657517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/17/657517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/17/657517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/17/657517/1"}}},"commit":{"parents":[{"commit":"a7380d4e8686794e45c7ec8dec6ecaf139da6628","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a7380d4e8686794e45c7ec8dec6ecaf139da6628"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-07 02:50:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-07 03:08:57.000000000","tz":0},"subject":"Skip image validation if failed","message":"Skip image validation if failed\n\nThis allows image validation to fail and continue lauching the\ncontainer. This is because image validation might depend on\nexternal image registry (i.e. DockerHub) which is not always stable.\n\nChange-Id: I3c65cdef17d8279233bfbb74a38c9db79b01a2de\nCloses-Bug: #1827968\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2b9c0581cce2e0b9d2272fa734c744e6dbea3d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/2b9c0581cce2e0b9d2272fa734c744e6dbea3d3e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
