)]}'
{"id":"openstack%2Fzun~657518","triplet_id":"openstack%2Fzun~stable%2Fqueens~I3c65cdef17d8279233bfbb74a38c9db79b01a2de","project":"openstack/zun","branch":"stable/queens","topic":"bug/1827968","hashtags":[],"change_id":"I3c65cdef17d8279233bfbb74a38c9db79b01a2de","subject":"Skip image validation if failed","status":"MERGED","created":"2019-05-07 03:11:06.000000000","updated":"2019-05-11 15:48:13.000000000","submitted":"2019-05-11 15:48:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657518-1557589693268-5c776a74","meta_rev_id":"2fb558846ba5edfa0caa6a679ae5e34854952ad4","_number":657518,"virtual_id_number":657518,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-05-11 15:48:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-05-11 15:00:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-05-11 15:00:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-11 15:48:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4fa6a37480429be0c952f8de4a72c0fcda9de73","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-07 03:11:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce62904b75e7a8988fe9884cc708c959e18a83c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 04:16:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/18/657518/1/check/openstack-tox-pep8/7f78da8/ : FAILURE in 5m 58s\n- openstack-tox-py27 http://logs.openstack.org/18/657518/1/check/openstack-tox-py27/2688232/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/18/657518/1/check/openstack-tox-py35/da7b586/ : SUCCESS in 5m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/657518/1/check/build-openstack-sphinx-docs/22e5179/html/ : SUCCESS in 4m 54s\n- zun-tempest-docker-sql http://logs.openstack.org/18/657518/1/check/zun-tempest-docker-sql/19e23db/ : SUCCESS in 46m 01s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/18/657518/1/check/zun-tempest-py35-docker-sql/996e8c4/ : SUCCESS in 45m 38s\n- zun-tempest-docker-etcd http://logs.openstack.org/18/657518/1/check/zun-tempest-docker-etcd/ef66a63/ : SUCCESS in 46m 59s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/18/657518/1/check/zun-tempest-multinode-docker-sql/90c8643/ : FAILURE in 58m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36328d5b26ac59a06aa404a0a022506f5bae4627","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-07 04:28:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e661b46308273fc38933f2151c63f376f5cee6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 05:43:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/657518/2/check/openstack-tox-pep8/72b3c92/ : SUCCESS in 6m 47s\n- openstack-tox-py27 http://logs.openstack.org/18/657518/2/check/openstack-tox-py27/801090c/ : SUCCESS in 5m 09s\n- openstack-tox-py35 http://logs.openstack.org/18/657518/2/check/openstack-tox-py35/786ebba/ : SUCCESS in 4m 49s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/657518/2/check/build-openstack-sphinx-docs/6b6393d/html/ : SUCCESS in 4m 49s\n- zun-tempest-docker-sql http://logs.openstack.org/18/657518/2/check/zun-tempest-docker-sql/4baaea6/ : SUCCESS in 45m 26s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/18/657518/2/check/zun-tempest-py35-docker-sql/22ece3d/ : SUCCESS in 46m 38s\n- zun-tempest-docker-etcd http://logs.openstack.org/18/657518/2/check/zun-tempest-docker-etcd/a0d3fb7/ : SUCCESS in 52m 22s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/18/657518/2/check/zun-tempest-multinode-docker-sql/362b87f/ : FAILURE in 1h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b0fbc7dd6771b54f40fec659e309eaa93793218","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-11 15:00:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9139d7e772f0723a0940ee81c915e6df340979bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 15:00:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f606de00615b692c73d5f932f94a85a6be4c68a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 15:48:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0ae3d6320e8f3a9586fa384b7b0dc899daf4ef95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 15:48:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/657518/2/gate/openstack-tox-pep8/54740db/ : SUCCESS in 8m 26s\n- openstack-tox-py27 http://logs.openstack.org/18/657518/2/gate/openstack-tox-py27/e0a8070/ : SUCCESS in 4m 53s\n- openstack-tox-py35 http://logs.openstack.org/18/657518/2/gate/openstack-tox-py35/fff0855/ : SUCCESS in 5m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/657518/2/gate/build-openstack-sphinx-docs/cfb3975/html/ : SUCCESS in 6m 04s\n- zun-tempest-docker-sql http://logs.openstack.org/18/657518/2/gate/zun-tempest-docker-sql/cffaee1/ : SUCCESS in 43m 30s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/18/657518/2/gate/zun-tempest-py35-docker-sql/257b787/ : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6556aef67c59c2a87302835140215e454639fa20","revisions":{"c2edc84b5806e9647a0af2d0b26ba39a94a5bdd2":{"kind":"REWORK","_number":1,"created":"2019-05-07 03:11:06.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/18/657518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/18/657518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/18/657518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/18/657518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/18/657518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/18/657518/1"}}},"commit":{"parents":[{"commit":"7cf655f4d9ab28ea3221c24200251611bde51c61","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7cf655f4d9ab28ea3221c24200251611bde51c61"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-07 02:50:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-07 03:11:06.000000000","tz":0},"subject":"Skip image validation if failed","message":"Skip image validation if failed\n\nThis allows image validation to fail and continue lauching the\ncontainer. This is because image validation might depend on\nexternal image registry (i.e. DockerHub) which is not always stable.\n\nChange-Id: I3c65cdef17d8279233bfbb74a38c9db79b01a2de\nCloses-Bug: #1827968\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c2edc84b5806e9647a0af2d0b26ba39a94a5bdd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/c2edc84b5806e9647a0af2d0b26ba39a94a5bdd2"}]},"branch":"refs/heads/stable/queens"},"6556aef67c59c2a87302835140215e454639fa20":{"kind":"REWORK","_number":2,"created":"2019-05-07 04:28:28.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/18/657518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/18/657518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/18/657518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/18/657518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/18/657518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/18/657518/2"}}},"commit":{"parents":[{"commit":"7cf655f4d9ab28ea3221c24200251611bde51c61","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/7cf655f4d9ab28ea3221c24200251611bde51c61"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-07 02:50:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-07 04:28:21.000000000","tz":0},"subject":"Skip image validation if failed","message":"Skip image validation if failed\n\nThis allows image validation to fail and continue lauching the\ncontainer. This is because image validation might depend on\nexternal image registry (i.e. DockerHub) which is not always stable.\n\nChange-Id: I3c65cdef17d8279233bfbb74a38c9db79b01a2de\nCloses-Bug: #1827968\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6556aef67c59c2a87302835140215e454639fa20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6556aef67c59c2a87302835140215e454639fa20"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
