)]}'
{"id":"openstack%2Fzun~658669","triplet_id":"openstack%2Fzun~master~Ife84a812875b6a3cf8569ab3435ce586dd7150df","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ife84a812875b6a3cf8569ab3435ce586dd7150df","subject":"Refactor get_available_resources","status":"MERGED","created":"2019-05-12 15:55:39.000000000","updated":"2019-05-17 02:00:45.000000000","submitted":"2019-05-17 02:00:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658669-1558058445528-4b33e095","meta_rev_id":"172425dbc643cd0d9b6e3a06270dba68fe1c72c7","_number":658669,"virtual_id_number":658669,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-05-17 02:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":2,"date":"2019-05-17 00:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":1,"date":"2019-05-17 00:55:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 00:55:24.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-05-17 02:00:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f87d3c95e6af85cebd18fc84c9f7dc9e17e787d3","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-12 15:55:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"970ed53e303a8c9a6644b3b6bad6c0dd2b24ebfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 17:46:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/69/658669/1/check/openstack-tox-cover/8d82ac7/ : FAILURE in 4m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/658669/1/check/openstack-tox-lower-constraints/8c75873/ : FAILURE in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/69/658669/1/check/openstack-tox-pep8/e514fad/ : SUCCESS in 5m 58s\n- openstack-tox-py27 http://logs.openstack.org/69/658669/1/check/openstack-tox-py27/aca4e4b/ : FAILURE in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/69/658669/1/check/openstack-tox-py36/150773b/ : FAILURE in 5m 44s\n- openstack-tox-py37 http://logs.openstack.org/69/658669/1/check/openstack-tox-py37/82cd3da/ : FAILURE in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/69/658669/1/check/openstack-tox-docs/6d242f7/html/ : SUCCESS in 3m 16s\n- zun-tempest-docker-sql http://logs.openstack.org/69/658669/1/check/zun-tempest-docker-sql/c788095/ : TIMED_OUT in 1h 15m 35s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/69/658669/1/check/zun-tempest-py3-docker-sql/15cf9d7/ : TIMED_OUT in 1h 13m 09s\n- zun-tempest-docker-etcd http://logs.openstack.org/69/658669/1/check/zun-tempest-docker-etcd/9a9290f/ : TIMED_OUT in 1h 12m 25s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/69/658669/1/check/zun-tempest-multinode-docker-sql/12ea5c3/ : FAILURE in 1h 48m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f8c6a1a40ee15267535bdfc50a728d9afce2b485","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-12 17:54:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"24f8e1f161d60877fc9b308a2f3abd0eb6643429","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 19:02:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/69/658669/2/check/openstack-tox-cover/895097d/cover/ : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/658669/2/check/openstack-tox-lower-constraints/1eb336e/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/69/658669/2/check/openstack-tox-pep8/3b4690d/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/69/658669/2/check/openstack-tox-py27/e5da655/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/69/658669/2/check/openstack-tox-py36/fb838d0/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/69/658669/2/check/openstack-tox-py37/5b5dee1/ : SUCCESS in 6m 13s\n- openstack-tox-docs http://logs.openstack.org/69/658669/2/check/openstack-tox-docs/e413488/html/ : SUCCESS in 3m 21s\n- zun-tempest-docker-sql http://logs.openstack.org/69/658669/2/check/zun-tempest-docker-sql/a5d9288/ : SUCCESS in 45m 46s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/69/658669/2/check/zun-tempest-py3-docker-sql/3ff058c/ : SUCCESS in 46m 41s\n- zun-tempest-docker-etcd http://logs.openstack.org/69/658669/2/check/zun-tempest-docker-etcd/e4db69a/ : SUCCESS in 47m 29s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/69/658669/2/check/zun-tempest-multinode-docker-sql/7182e9b/ : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":2},{"id":"4926fc19bd114b73b0206c11e9388c34dabcb45a","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-05-17 00:55:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4889b12704b688f097e55fee0eb411ad49441df3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 00:55:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"13018a50bd658a85a07b834bc55cd51e4f3b03f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:00:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/69/658669/2/gate/openstack-tox-lower-constraints/ef391e0/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/69/658669/2/gate/openstack-tox-pep8/177d92c/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/69/658669/2/gate/openstack-tox-py27/6c3b84a/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/69/658669/2/gate/openstack-tox-py36/3945631/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/69/658669/2/gate/openstack-tox-py37/0084158/ : SUCCESS in 4m 22s\n- openstack-tox-docs http://logs.openstack.org/69/658669/2/gate/openstack-tox-docs/423df64/html/ : SUCCESS in 3m 25s\n- zun-tempest-docker-sql http://logs.openstack.org/69/658669/2/gate/zun-tempest-docker-sql/f6a6c35/ : SUCCESS in 48m 35s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/69/658669/2/gate/zun-tempest-py3-docker-sql/72dd8fa/ : SUCCESS in 47m 20s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/69/658669/2/gate/zun-tempest-multinode-docker-sql/fa93e7f/ : SUCCESS in 56m 39s","accounts_in_message":[],"_revision_number":2},{"id":"172425dbc643cd0d9b6e3a06270dba68fe1c72c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:00:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9776a098b27a4a2d28764a75f01923bdd3284054","revisions":{"6ac179ec682377c1ea4d48327aa5e7b8c7b6aad9":{"kind":"REWORK","_number":1,"created":"2019-05-12 15:55:39.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/69/658669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/69/658669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/69/658669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/69/658669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/69/658669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/69/658669/1"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-12 15:55:10.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-12 15:55:10.000000000","tz":0},"subject":"[WIP] Refactor get_available_resources","message":"[WIP] Refactor get_available_resources\n\nChange-Id: Ife84a812875b6a3cf8569ab3435ce586dd7150df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6ac179ec682377c1ea4d48327aa5e7b8c7b6aad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6ac179ec682377c1ea4d48327aa5e7b8c7b6aad9"}]},"branch":"refs/heads/master"},"9776a098b27a4a2d28764a75f01923bdd3284054":{"kind":"REWORK","_number":2,"created":"2019-05-12 17:54:04.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/69/658669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/69/658669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/69/658669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/69/658669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/69/658669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/69/658669/2"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-12 15:55:10.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-12 17:53:37.000000000","tz":0},"subject":"Refactor get_available_resources","message":"Refactor get_available_resources\n\nChange-Id: Ife84a812875b6a3cf8569ab3435ce586dd7150df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9776a098b27a4a2d28764a75f01923bdd3284054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9776a098b27a4a2d28764a75f01923bdd3284054"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
