)]}'
{"id":"openstack%2Fzun~658690","triplet_id":"openstack%2Fzun~master~I21a4ab8934196afd43ae15f5aa76464dab2b5bd2","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I21a4ab8934196afd43ae15f5aa76464dab2b5bd2","subject":"Remove get_cpu_used from container driver","status":"MERGED","created":"2019-05-12 18:50:15.000000000","updated":"2019-05-27 03:23:55.000000000","submitted":"2019-05-27 03:23:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658690-1558927435526-3080c1e4","meta_rev_id":"fe6bc655d834c84cb8147b98aabb83f6bb6970c2","_number":658690,"virtual_id_number":658690,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-05-27 03:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":2,"date":"2019-05-27 02:16:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":1,"date":"2019-05-27 02:16:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-27 02:16:16.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-05-27 03:23:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"379ef7f97039853ed7d2d17a0ea7ed12d5308d3b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-12 18:50:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c537be0b0f980cbc19d2be15246ecb65c9a5e96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 19:56:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/90/658690/1/check/openstack-tox-cover/9d04020/cover/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/658690/1/check/openstack-tox-lower-constraints/a20a47f/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/90/658690/1/check/openstack-tox-pep8/4595bd7/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/90/658690/1/check/openstack-tox-py27/82fbd6b/ : SUCCESS in 5m 21s\n- openstack-tox-py36 http://logs.openstack.org/90/658690/1/check/openstack-tox-py36/1bf8dd1/ : SUCCESS in 5m 04s\n- openstack-tox-py37 http://logs.openstack.org/90/658690/1/check/openstack-tox-py37/8ef7c8d/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/90/658690/1/check/openstack-tox-docs/7bedc74/html/ : SUCCESS in 4m 28s\n- zun-tempest-docker-sql http://logs.openstack.org/90/658690/1/check/zun-tempest-docker-sql/cc0402a/ : SUCCESS in 55m 29s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/90/658690/1/check/zun-tempest-py3-docker-sql/d806456/ : SUCCESS in 46m 39s\n- zun-tempest-docker-etcd http://logs.openstack.org/90/658690/1/check/zun-tempest-docker-etcd/0618540/ : SUCCESS in 51m 40s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/90/658690/1/check/zun-tempest-multinode-docker-sql/bb39866/ : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":1},{"id":"480b65d17ffbccec65dadd93d30af6ea8b25f7b1","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-05-13 06:56:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0777b3da232a990a78896b870cca85f3282cfa49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:00:46.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"30afe64bee2fb8e5b3d8b70b39a0536e6a98b07e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 16:08:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8160ac2dfa1699d0d5ada830d972872e567b72f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 17:21:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/90/658690/2/check/openstack-tox-cover/30f3f21/cover/ : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/658690/2/check/openstack-tox-lower-constraints/a194a44/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/90/658690/2/check/openstack-tox-pep8/032fb08/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/90/658690/2/check/openstack-tox-py27/f52a5b5/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/90/658690/2/check/openstack-tox-py36/518c7db/ : SUCCESS in 4m 12s\n- openstack-tox-py37 http://logs.openstack.org/90/658690/2/check/openstack-tox-py37/63d426d/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/90/658690/2/check/openstack-tox-docs/653c0b9/html/ : SUCCESS in 3m 15s\n- zun-tempest-docker-sql http://logs.openstack.org/90/658690/2/check/zun-tempest-docker-sql/1d0aea3/ : SUCCESS in 52m 17s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/90/658690/2/check/zun-tempest-py3-docker-sql/6630965/ : SUCCESS in 50m 02s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/90/658690/2/check/zun-tempest-multinode-docker-sql/855ae08/ : SUCCESS in 1h 07m 06s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/90/658690/2/check/kolla-ansible-ubuntu-source-zun/7a4f824/ : SUCCESS in 50m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3efc3d01371ef06f2da54390f679eeef170c185","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-05-27 02:16:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81d06781a4641ccc27dd49eff50f53a88d24a3ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 02:16:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4382358c533e53af42e36b232b2f32de6f103cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 03:23:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fe6bc655d834c84cb8147b98aabb83f6bb6970c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 03:23:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/658690/2/gate/openstack-tox-lower-constraints/5b6329b/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/90/658690/2/gate/openstack-tox-pep8/b54f725/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/90/658690/2/gate/openstack-tox-py27/cf6a037/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/90/658690/2/gate/openstack-tox-py36/79573a9/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/90/658690/2/gate/openstack-tox-py37/eef7697/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/90/658690/2/gate/openstack-tox-docs/b86d7e2/html/ : SUCCESS in 3m 30s\n- zun-tempest-docker-sql http://logs.openstack.org/90/658690/2/gate/zun-tempest-docker-sql/f370aca/ : SUCCESS in 50m 15s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/90/658690/2/gate/zun-tempest-py3-docker-sql/57b24d9/ : SUCCESS in 41m 53s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/90/658690/2/gate/zun-tempest-multinode-docker-sql/b8b7b9a/ : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"37ba05927c2d076fbe4854a65fe3a8f47faa4fb5","revisions":{"5c10e3a9a3b03827664baa793aa8134aef25cb36":{"kind":"REWORK","_number":1,"created":"2019-05-12 18:50:15.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/90/658690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/90/658690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/658690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/658690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/658690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/90/658690/1"}}},"commit":{"parents":[{"commit":"9776a098b27a4a2d28764a75f01923bdd3284054","subject":"Refactor get_available_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9776a098b27a4a2d28764a75f01923bdd3284054"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-12 18:42:24.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-12 18:50:06.000000000","tz":0},"subject":"Remove get_cpu_used from container driver","message":"Remove get_cpu_used from container driver\n\nThis method is heavy since it inspects each container in Docker.\nThe computed value is basically not used. This commit removes it.\n\nChange-Id: I21a4ab8934196afd43ae15f5aa76464dab2b5bd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5c10e3a9a3b03827664baa793aa8134aef25cb36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5c10e3a9a3b03827664baa793aa8134aef25cb36"}]},"branch":"refs/heads/master"},"37ba05927c2d076fbe4854a65fe3a8f47faa4fb5":{"kind":"REWORK","_number":2,"created":"2019-05-18 16:08:50.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/90/658690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/90/658690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/658690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/658690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/90/658690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/90/658690/2"}}},"commit":{"parents":[{"commit":"d1684a67950b6dcdaca013a0c1a920b76aaed3f8","subject":"Merge \"Remove etcd db backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1684a67950b6dcdaca013a0c1a920b76aaed3f8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-12 18:42:24.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-18 16:08:44.000000000","tz":0},"subject":"Remove get_cpu_used from container driver","message":"Remove get_cpu_used from container driver\n\nThis method is heavy since it inspects each container in Docker.\nThe computed value is basically not used. This commit removes it.\n\nChange-Id: I21a4ab8934196afd43ae15f5aa76464dab2b5bd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/37ba05927c2d076fbe4854a65fe3a8f47faa4fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/37ba05927c2d076fbe4854a65fe3a8f47faa4fb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
