)]}'
{"id":"openstack%2Fzun~658722","triplet_id":"openstack%2Fzun~master~I298c1e9dfdab01b55504895b6279437089014c08","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I298c1e9dfdab01b55504895b6279437089014c08","subject":"api-ref: add API reference for capsule","status":"MERGED","created":"2019-05-13 03:31:40.000000000","updated":"2019-05-22 02:32:13.000000000","submitted":"2019-05-22 02:32:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658722-1558492333378-bd4b8c07","meta_rev_id":"af6fed39f9e435b253ed287a22bb6823de232ea9","_number":658722,"virtual_id_number":658722,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-20 01:11:54.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-05-22 02:32:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-05-22 02:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-05-22 02:23:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 02:23:04.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-05-22 02:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42e77f712b555b1a930db1fbdc310a01bb823499","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-13 03:31:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be5e5d31dea7c05c24a21bfcd5ede6fd71c7cd8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 04:35:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/22/658722/1/check/build-openstack-api-ref/35f7644/html/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/22/658722/1/check/openstack-tox-pep8/8b17478/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/22/658722/1/check/openstack-tox-docs/351f0fd/html/ : SUCCESS in 4m 29s\n- zun-tempest-docker-sql http://logs.openstack.org/22/658722/1/check/zun-tempest-docker-sql/296b404/ : SUCCESS in 50m 16s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/22/658722/1/check/zun-tempest-py3-docker-sql/d546a5a/ : SUCCESS in 47m 00s\n- zun-tempest-docker-etcd http://logs.openstack.org/22/658722/1/check/zun-tempest-docker-etcd/27b0682/ : SUCCESS in 48m 35s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/22/658722/1/check/zun-tempest-multinode-docker-sql/0466ac5/ : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":1},{"id":"57d24a14d7746121c44e9e2099ae884509882ad4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-15 03:18:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffa6b5241c576bda54e5acfc701ca863dfd34df4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 04:25:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/22/658722/2/check/build-openstack-api-ref/376b84d/ : FAILURE in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/22/658722/2/check/openstack-tox-pep8/d22110f/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/22/658722/2/check/openstack-tox-docs/9ed2b14/html/ : SUCCESS in 3m 24s\n- zun-tempest-docker-sql http://logs.openstack.org/22/658722/2/check/zun-tempest-docker-sql/29c777a/ : SUCCESS in 53m 38s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/22/658722/2/check/zun-tempest-py3-docker-sql/9937df5/ : SUCCESS in 45m 48s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/22/658722/2/check/zun-tempest-multinode-docker-sql/690851b/ : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3ed258a741e8f984e8adbce5c606943a601b6b1b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-16 03:54:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e762a73b2abf58543137965ef2ab54f6cf4ab9c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:09:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/22/658722/3/check/build-openstack-api-ref/3d5fe3f/ : FAILURE in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/22/658722/3/check/openstack-tox-pep8/0e7e786/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/22/658722/3/check/openstack-tox-docs/ab4905a/html/ : SUCCESS in 4m 12s\n- zun-tempest-docker-sql http://logs.openstack.org/22/658722/3/check/zun-tempest-docker-sql/a77c84a/ : SUCCESS in 48m 42s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/22/658722/3/check/zun-tempest-py3-docker-sql/40a9931/ : SUCCESS in 51m 11s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/22/658722/3/check/zun-tempest-multinode-docker-sql/89d84e6/ : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":3},{"id":"48d6acdb1c7e587bd50cacb92bb7f1e234f6916d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-17 02:12:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"81384c0c80abcdd43fffe58aaf660562fa25ffb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 03:28:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/22/658722/4/check/build-openstack-api-ref/10967fc/ : FAILURE in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/22/658722/4/check/openstack-tox-pep8/89f6621/ : SUCCESS in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/22/658722/4/check/openstack-tox-docs/a1efa17/html/ : SUCCESS in 3m 22s\n- zun-tempest-docker-sql http://logs.openstack.org/22/658722/4/check/zun-tempest-docker-sql/17d766e/ : SUCCESS in 46m 19s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/22/658722/4/check/zun-tempest-py3-docker-sql/82f5153/ : SUCCESS in 46m 35s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/22/658722/4/check/zun-tempest-multinode-docker-sql/63c1e5b/ : SUCCESS in 1h 13m 28s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/22/658722/4/check/kolla-ansible-ubuntu-source-zun/fb8520b/ : SUCCESS in 42m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0145a0d929aa0706912ee2fd0ca39d7df06c4861","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 19:26:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"46002848e9c28cd700e5b73c731de20dbff53e63","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 19:46:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b1314499d808ed93b820c4a76fb541b84b81f443","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 20:51:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/22/658722/6/check/build-openstack-api-ref/06f06e5/ : FAILURE in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/22/658722/6/check/openstack-tox-pep8/8ce1b7c/ : SUCCESS in 4m 04s\n- openstack-tox-docs http://logs.openstack.org/22/658722/6/check/openstack-tox-docs/886ae77/html/ : SUCCESS in 3m 11s\n- zun-tempest-docker-sql http://logs.openstack.org/22/658722/6/check/zun-tempest-docker-sql/ddd252f/ : SUCCESS in 51m 05s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/22/658722/6/check/zun-tempest-py3-docker-sql/25fa8e6/ : SUCCESS in 49m 52s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/22/658722/6/check/zun-tempest-multinode-docker-sql/48299ca/ : SUCCESS in 58m 51s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/22/658722/6/check/kolla-ansible-ubuntu-source-zun/8b4c382/ : SUCCESS in 40m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d8893f5f8bf3cbff527f5d4cb083ab8d79925d89","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 23:01:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e3fcf6f4ee4c0487ef0551f754758111fbd70cb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 00:00:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/22/658722/7/check/build-openstack-api-ref/1239981/ : FAILURE in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/22/658722/7/check/openstack-tox-pep8/a5f825a/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/22/658722/7/check/openstack-tox-docs/5ba4246/html/ : SUCCESS in 2m 53s\n- zun-tempest-docker-sql http://logs.openstack.org/22/658722/7/check/zun-tempest-docker-sql/caa8852/ : SUCCESS in 39m 51s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/22/658722/7/check/zun-tempest-py3-docker-sql/48690e9/ : SUCCESS in 46m 38s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/22/658722/7/check/zun-tempest-multinode-docker-sql/0dd6cb4/ : SUCCESS in 57m 27s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/22/658722/7/check/kolla-ansible-ubuntu-source-zun/c58158c/ : SUCCESS in 42m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e591b5e63b0a5b62fb62f2303a45342b5b283955","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-19 18:49:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"68a5808bfb04153b832245e04a699e9f8c73e551","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-19 19:22:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f6eaa32e099609aa6d69079ce5c636bc9caa2a77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 20:26:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/22/658722/9/check/build-openstack-api-ref/b59c380/html/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/22/658722/9/check/openstack-tox-pep8/f5adc31/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/22/658722/9/check/openstack-tox-docs/1aa7ae2/html/ : SUCCESS in 3m 19s\n- zun-tempest-docker-sql http://logs.openstack.org/22/658722/9/check/zun-tempest-docker-sql/a18076b/ : SUCCESS in 49m 02s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/22/658722/9/check/zun-tempest-py3-docker-sql/e29c1e1/ : SUCCESS in 50m 44s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/22/658722/9/check/zun-tempest-multinode-docker-sql/2865e6b/ : SUCCESS in 58m 51s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/22/658722/9/check/kolla-ansible-ubuntu-source-zun/be00d27/ : SUCCESS in 39m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"985bf7452464a4e6be10b6acd16f4d5e39bcc95f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-19 21:17:40.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"5a930f0ca11b6955822cd8776d05df6f0b9f018d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 22:24:49.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/22/658722/10/check/build-openstack-api-ref/227dcf0/html/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/22/658722/10/check/openstack-tox-pep8/315c0cc/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/22/658722/10/check/openstack-tox-docs/7eb2c71/html/ : SUCCESS in 3m 11s\n- zun-tempest-docker-sql http://logs.openstack.org/22/658722/10/check/zun-tempest-docker-sql/db3f612/ : FAILURE in 48m 47s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/22/658722/10/check/zun-tempest-py3-docker-sql/64ced0e/ : SUCCESS in 46m 55s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/22/658722/10/check/zun-tempest-multinode-docker-sql/70d514a/ : SUCCESS in 1h 01m 17s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/22/658722/10/check/kolla-ansible-ubuntu-source-zun/b0b9f19/ : SUCCESS in 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fb06d95fe716dbe3168153c22874746bfb950bfc","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-20 01:11:54.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"88acfa53b98d93505d5bc83c38290da24375a702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 02:15:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/22/658722/10/check/build-openstack-api-ref/9b0e4f7/html/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/22/658722/10/check/openstack-tox-pep8/df2b66e/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/22/658722/10/check/openstack-tox-docs/c40628f/html/ : SUCCESS in 4m 00s\n- zun-tempest-docker-sql http://logs.openstack.org/22/658722/10/check/zun-tempest-docker-sql/b3b26e0/ : SUCCESS in 41m 22s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/22/658722/10/check/zun-tempest-py3-docker-sql/7a91830/ : SUCCESS in 49m 10s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/22/658722/10/check/zun-tempest-multinode-docker-sql/24932c9/ : SUCCESS in 1h 01m 39s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/22/658722/10/check/kolla-ansible-ubuntu-source-zun/e84d447/ : SUCCESS in 43m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3438a8f49c5f14dce77839efc55b34b8b509fc64","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-05-22 02:23:04.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"413cc3b7c65ca4c339a1ef6ce12b1359fbffbe89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 02:23:16.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"a42d0d83b393005fba243abdcd438b77839e4149","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 02:32:13.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/22/658722/10/gate/build-openstack-api-ref/e55508c/html/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/22/658722/10/gate/openstack-tox-pep8/70cff8b/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/22/658722/10/gate/openstack-tox-docs/84523cd/html/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":10},{"id":"af6fed39f9e435b253ed287a22bb6823de232ea9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 02:32:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"125f2a25b6f226df4b955c6d9364cc022410e39a","revisions":{"556c65cbdc10639651352e0f4b0f162bb493f746":{"kind":"REWORK","_number":1,"created":"2019-05-13 03:31:40.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/658722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/658722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/658722/1"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"subject":"[WIP] Add API reference for capsule","message":"[WIP] Add API reference for capsule\n\nChange-Id: I298c1e9dfdab01b55504895b6279437089014c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/556c65cbdc10639651352e0f4b0f162bb493f746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/556c65cbdc10639651352e0f4b0f162bb493f746"}]},"branch":"refs/heads/master"},"0ce20fa0f849d8d9baa7a8eaa06b9a377172d4c6":{"kind":"REWORK","_number":2,"created":"2019-05-15 03:18:29.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/658722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/658722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/658722/2"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-15 03:18:19.000000000","tz":0},"subject":"[WIP] Add API reference for capsule","message":"[WIP] Add API reference for capsule\n\nChange-Id: I298c1e9dfdab01b55504895b6279437089014c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0ce20fa0f849d8d9baa7a8eaa06b9a377172d4c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/0ce20fa0f849d8d9baa7a8eaa06b9a377172d4c6"}]},"branch":"refs/heads/master"},"a8ae64240ad274efa8c42ab9ade07803d7805fea":{"kind":"REWORK","_number":3,"created":"2019-05-16 03:54:56.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/658722/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/658722/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/658722/3"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-16 03:54:46.000000000","tz":0},"subject":"[WIP] Add API reference for capsule","message":"[WIP] Add API reference for capsule\n\nChange-Id: I298c1e9dfdab01b55504895b6279437089014c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a8ae64240ad274efa8c42ab9ade07803d7805fea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/a8ae64240ad274efa8c42ab9ade07803d7805fea"}]},"branch":"refs/heads/master"},"80bbca01055c1360b0475337ad9b50ad8405f5fe":{"kind":"REWORK","_number":4,"created":"2019-05-17 02:12:34.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/658722/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/658722/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/658722/4"}}},"commit":{"parents":[{"commit":"b00bba2483a372c76c2d64c06c701a7c8d9e3a7b","subject":"Disable image validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/b00bba2483a372c76c2d64c06c701a7c8d9e3a7b"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-17 02:12:21.000000000","tz":0},"subject":"[WIP] Add API reference for capsule","message":"[WIP] Add API reference for capsule\n\nChange-Id: I298c1e9dfdab01b55504895b6279437089014c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/80bbca01055c1360b0475337ad9b50ad8405f5fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/80bbca01055c1360b0475337ad9b50ad8405f5fe"}]},"branch":"refs/heads/master"},"59bf8eaea307cede627551ed7dc40cecd7bfdf4d":{"kind":"REWORK","_number":5,"created":"2019-05-18 19:26:44.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/658722/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/658722/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/658722/5"}}},"commit":{"parents":[{"commit":"d1684a67950b6dcdaca013a0c1a920b76aaed3f8","subject":"Merge \"Remove etcd db backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1684a67950b6dcdaca013a0c1a920b76aaed3f8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-18 19:26:38.000000000","tz":0},"subject":"[WIP] Add API reference for capsule","message":"[WIP] Add API reference for capsule\n\nChange-Id: I298c1e9dfdab01b55504895b6279437089014c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/59bf8eaea307cede627551ed7dc40cecd7bfdf4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/59bf8eaea307cede627551ed7dc40cecd7bfdf4d"}]},"branch":"refs/heads/master"},"d5e2bbf86ff269725b74eedb2cd58ce311bd738d":{"kind":"REWORK","_number":6,"created":"2019-05-18 19:46:31.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/658722/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/658722/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/658722/6"}}},"commit":{"parents":[{"commit":"d1684a67950b6dcdaca013a0c1a920b76aaed3f8","subject":"Merge \"Remove etcd db backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1684a67950b6dcdaca013a0c1a920b76aaed3f8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-18 19:46:25.000000000","tz":0},"subject":"[WIP] Add API reference for capsule","message":"[WIP] Add API reference for capsule\n\nChange-Id: I298c1e9dfdab01b55504895b6279437089014c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d5e2bbf86ff269725b74eedb2cd58ce311bd738d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d5e2bbf86ff269725b74eedb2cd58ce311bd738d"}]},"branch":"refs/heads/master"},"d1058588ab30d1a43cc7437db0d835b64716e33b":{"kind":"REWORK","_number":7,"created":"2019-05-18 23:01:44.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/658722/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/658722/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/658722/7"}}},"commit":{"parents":[{"commit":"d1684a67950b6dcdaca013a0c1a920b76aaed3f8","subject":"Merge \"Remove etcd db backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1684a67950b6dcdaca013a0c1a920b76aaed3f8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-18 23:01:37.000000000","tz":0},"subject":"[WIP] Add API reference for capsule","message":"[WIP] Add API reference for capsule\n\nChange-Id: I298c1e9dfdab01b55504895b6279437089014c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1058588ab30d1a43cc7437db0d835b64716e33b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1058588ab30d1a43cc7437db0d835b64716e33b"}]},"branch":"refs/heads/master"},"acd7cbd37510fa80bcef2173281f6a2a243b2d4a":{"kind":"REWORK","_number":8,"created":"2019-05-19 18:49:31.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/658722/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/658722/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/658722/8"}}},"commit":{"parents":[{"commit":"d1684a67950b6dcdaca013a0c1a920b76aaed3f8","subject":"Merge \"Remove etcd db backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1684a67950b6dcdaca013a0c1a920b76aaed3f8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-19 18:49:22.000000000","tz":0},"subject":"[WIP] Add API reference for capsule","message":"[WIP] Add API reference for capsule\n\nChange-Id: I298c1e9dfdab01b55504895b6279437089014c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/acd7cbd37510fa80bcef2173281f6a2a243b2d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/acd7cbd37510fa80bcef2173281f6a2a243b2d4a"}]},"branch":"refs/heads/master"},"02843ce86ef3f77b763b18a0005204a012d3d913":{"kind":"REWORK","_number":9,"created":"2019-05-19 19:22:42.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/658722/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/658722/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/658722/9"}}},"commit":{"parents":[{"commit":"d1684a67950b6dcdaca013a0c1a920b76aaed3f8","subject":"Merge \"Remove etcd db backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1684a67950b6dcdaca013a0c1a920b76aaed3f8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-19 19:22:30.000000000","tz":0},"subject":"Add API reference for capsule","message":"Add API reference for capsule\n\nChange-Id: I298c1e9dfdab01b55504895b6279437089014c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/02843ce86ef3f77b763b18a0005204a012d3d913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/02843ce86ef3f77b763b18a0005204a012d3d913"}]},"branch":"refs/heads/master"},"125f2a25b6f226df4b955c6d9364cc022410e39a":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-05-19 21:17:40.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/658722/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/658722/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/658722/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/658722/10"}}},"commit":{"parents":[{"commit":"d1684a67950b6dcdaca013a0c1a920b76aaed3f8","subject":"Merge \"Remove etcd db backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1684a67950b6dcdaca013a0c1a920b76aaed3f8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 03:31:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-19 21:17:19.000000000","tz":0},"subject":"api-ref: add API reference for capsule","message":"api-ref: add API reference for capsule\n\nChange-Id: I298c1e9dfdab01b55504895b6279437089014c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/125f2a25b6f226df4b955c6d9364cc022410e39a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/125f2a25b6f226df4b955c6d9364cc022410e39a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
