)]}'
{"id":"openstack%2Fzun~659922","triplet_id":"openstack%2Fzun~stable%2Fqueens~I54bf68a6d9f3d79f47fbcefa55509bc0993027bb","project":"openstack/zun","branch":"stable/queens","hashtags":[],"change_id":"I54bf68a6d9f3d79f47fbcefa55509bc0993027bb","subject":"Trivial: fix a format issue in api history docs","status":"MERGED","created":"2019-05-18 18:29:00.000000000","updated":"2019-05-18 20:44:21.000000000","submitted":"2019-05-18 18:49:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659922-1558205348295-38c90c1d","meta_rev_id":"a8f1af9b62ff2daf33d26322d43c282b60339c05","_number":659922,"virtual_id_number":659922,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-05-18 18:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-05-18 18:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-05-18 18:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-18 18:49:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b13039ead90972406d69177cda03bdbf78e94a28","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 18:29:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"ca7d71be539411d8352064b1b18f4090ef0150ad","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 18:29:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1834d8fed55954c8dc4973222bdab2f2e785355","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 18:39:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/659922/1/check/openstack-tox-pep8/0a95cec/ : SUCCESS in 9m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/22/659922/1/check/build-openstack-sphinx-docs/f9f2059/html/ : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0874e1b853a88b7514cd67ec97856dcda128bcd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 18:39:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9aa9625053e9c31629cde0852db17ba0e662c3e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 18:49:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/659922/1/gate/openstack-tox-pep8/ad7cc69/ : SUCCESS in 9m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/22/659922/1/gate/build-openstack-sphinx-docs/d7ad640/html/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a6a9c8197064e29ca1692e470e94229252442bd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 18:49:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a8f1af9b62ff2daf33d26322d43c282b60339c05","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 20:44:21.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit 4a9097ca3c2d2d3f5b0746b9f56dab177c16e09b","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"becf0efcbc80125b3f63474fd760e2d529eb1433","revisions":{"becf0efcbc80125b3f63474fd760e2d529eb1433":{"kind":"REWORK","_number":1,"created":"2019-05-18 18:29:00.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/659922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/22/659922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/659922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/659922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/22/659922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/22/659922/1"}}},"commit":{"parents":[{"commit":"6556aef67c59c2a87302835140215e454639fa20","subject":"Skip image validation if failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6556aef67c59c2a87302835140215e454639fa20"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-05 22:54:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-18 18:29:00.000000000","tz":0},"subject":"Trivial: fix a format issue in api history docs","message":"Trivial: fix a format issue in api history docs\n\nChange-Id: I54bf68a6d9f3d79f47fbcefa55509bc0993027bb\n(cherry picked from commit d62f9438ce7a1edaa29cdf22d360adfafed5c230)\n(cherry picked from commit 9bfbbd9a4b09c808305191de630ee531f3d8ca9f)\n(cherry picked from commit 6a523e4128b647832d92deca8cd85f2bb0a7ba53)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/becf0efcbc80125b3f63474fd760e2d529eb1433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/becf0efcbc80125b3f63474fd760e2d529eb1433"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
