)]}'
{"id":"openstack%2Fzun~659923","triplet_id":"openstack%2Fzun~stable%2Fstein~Iba9ab6c50565e470bdb3fc8d64c59c5bebd6fd36","project":"openstack/zun","branch":"stable/stein","hashtags":[],"change_id":"Iba9ab6c50565e470bdb3fc8d64c59c5bebd6fd36","subject":"Fix requirement job","status":"MERGED","created":"2019-05-18 18:30:01.000000000","updated":"2019-05-18 20:43:54.000000000","submitted":"2019-05-18 19:57:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659923-1558209447325-354c625b","meta_rev_id":"b25a2689bed70b939daca02173271e1f22098dc9","_number":659923,"virtual_id_number":659923,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-05-18 19:57:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-05-18 19:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-05-18 19:47:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-18 19:57:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54c344f69c34e81023cae5f0a353ae5bfbd035d0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 18:30:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a80d9e34f9dda0bff461766b8869d86f9d601f76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 19:03:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/23/659923/1/check/requirements-check/8aa9513/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/23/659923/1/check/openstack-tox-pep8/0338ed1/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/23/659923/1/check/openstack-tox-docs/1ed6f57/html/ : SUCCESS in 3m 14s\n- build-openstack-releasenotes http://logs.openstack.org/23/659923/1/check/build-openstack-releasenotes/c082290/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"57981ef1024f04ae6b272c274d257325ef5bd5a7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 19:47:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a1f49369ffa557f5d033286b3139b20432a3c9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 19:48:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc14bfad733b3870a4adde5ceb4bbedaf31e418e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 19:57:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/23/659923/1/gate/requirements-check/2351a69/ : SUCCESS in 2m 34s\n- openstack-tox-pep8 http://logs.openstack.org/23/659923/1/gate/openstack-tox-pep8/3b7e49c/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/23/659923/1/gate/openstack-tox-docs/7d16a4e/html/ : SUCCESS in 3m 05s\n- build-openstack-releasenotes http://logs.openstack.org/23/659923/1/gate/build-openstack-releasenotes/a9c1763/html/ : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6556f9ee0f824533b19f9b930a6f3c9aa5f67616","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 19:57:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b25a2689bed70b939daca02173271e1f22098dc9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-18 20:43:54.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 8961b948c3fc842dc8a582c5d8f2faa4d04e15c4","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59b45b34dda88818c31d1e7d99fe327d00c8f9fb","revisions":{"59b45b34dda88818c31d1e7d99fe327d00c8f9fb":{"kind":"REWORK","_number":1,"created":"2019-05-18 18:30:01.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/23/659923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/23/659923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/23/659923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/23/659923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/23/659923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/23/659923/1"}}},"commit":{"parents":[{"commit":"9bfbbd9a4b09c808305191de630ee531f3d8ca9f","subject":"Trivial: fix a format issue in api history docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/9bfbbd9a4b09c808305191de630ee531f3d8ca9f"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-13 02:10:36.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-18 18:30:01.000000000","tz":0},"subject":"Fix requirement job","message":"Fix requirement job\n\nSimilarly sphinx 2.0.0 drop py27 support do express that in\nglobal-requirements.\n\nChange-Id: Iba9ab6c50565e470bdb3fc8d64c59c5bebd6fd36\nSee: https://review.opendev.org/#/c/658634/\n(cherry picked from commit cb33446018b86fe8ef8577047d17d314fe4e7fc3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/59b45b34dda88818c31d1e7d99fe327d00c8f9fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/59b45b34dda88818c31d1e7d99fe327d00c8f9fb"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
