)]}'
{"id":"openstack%2Fzun~659987","triplet_id":"openstack%2Fzun~master~Ica9d392be3f12401f7ba51baffb0430bf92ac519","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Ica9d392be3f12401f7ba51baffb0430bf92ac519","subject":"Skip tempest jobs on api-ref changes","status":"MERGED","created":"2019-05-19 21:19:57.000000000","updated":"2019-05-22 02:32:11.000000000","submitted":"2019-05-22 02:32:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659987-1558492331711-19075ab0","meta_rev_id":"3e5489fd2260c69ebffc47a54af3796114e2b95f","_number":659987,"virtual_id_number":659987,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-05-22 02:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":2,"date":"2019-05-22 01:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":1,"date":"2019-05-22 01:37:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 01:37:40.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-05-22 02:32:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8eca16780a5d17608da9019f7c4cd6fc9644c1cc","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-19 21:19:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3751dab49048a057479364de9a609882888b02c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 22:23:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/87/659987/1/check/openstack-tox-cover/c35ddb1/cover/ : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/659987/1/check/openstack-tox-lower-constraints/e6d1ad4/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/87/659987/1/check/openstack-tox-pep8/3d17561/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/87/659987/1/check/openstack-tox-py27/5cda3c5/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/87/659987/1/check/openstack-tox-py36/79e02d0/ : SUCCESS in 4m 00s\n- openstack-tox-py37 http://logs.openstack.org/87/659987/1/check/openstack-tox-py37/97ce146/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/87/659987/1/check/openstack-tox-docs/e75ce67/html/ : SUCCESS in 2m 57s\n- zun-tempest-docker-sql http://logs.openstack.org/87/659987/1/check/zun-tempest-docker-sql/d9c2fb9/ : SUCCESS in 48m 54s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/87/659987/1/check/zun-tempest-py3-docker-sql/7d63a49/ : SUCCESS in 47m 44s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/87/659987/1/check/zun-tempest-multinode-docker-sql/4b27bb2/ : SUCCESS in 57m 58s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/87/659987/1/check/kolla-ansible-ubuntu-source-zun/b506268/ : SUCCESS in 39m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"989db66390b26d4d6259fbcb8f21cd6dadf4ff4b","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-05-22 01:37:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee513f6c926ff0d49e58f6d3271680e0558fc77e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 01:37:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d3fe9a09656b11f5ea156ba63f591210183f68e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 02:32:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e5489fd2260c69ebffc47a54af3796114e2b95f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 02:32:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/87/659987/1/gate/openstack-tox-lower-constraints/2ea678a/ : SUCCESS in 3m 55s\n- openstack-tox-pep8 http://logs.openstack.org/87/659987/1/gate/openstack-tox-pep8/ea17b42/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/87/659987/1/gate/openstack-tox-py27/4e32d2a/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/87/659987/1/gate/openstack-tox-py36/090daac/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/87/659987/1/gate/openstack-tox-py37/1b3d984/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/87/659987/1/gate/openstack-tox-docs/61db665/html/ : SUCCESS in 3m 19s\n- zun-tempest-docker-sql http://logs.openstack.org/87/659987/1/gate/zun-tempest-docker-sql/65544e6/ : SUCCESS in 40m 44s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/87/659987/1/gate/zun-tempest-py3-docker-sql/253f76d/ : SUCCESS in 46m 26s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/87/659987/1/gate/zun-tempest-multinode-docker-sql/6ca08f1/ : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56383590e74e16f9fdc8ee4b22e9cf3826cf4019","revisions":{"56383590e74e16f9fdc8ee4b22e9cf3826cf4019":{"kind":"REWORK","_number":1,"created":"2019-05-19 21:19:57.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/87/659987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/87/659987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/87/659987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/87/659987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/87/659987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/87/659987/1"}}},"commit":{"parents":[{"commit":"d1684a67950b6dcdaca013a0c1a920b76aaed3f8","subject":"Merge \"Remove etcd db backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/d1684a67950b6dcdaca013a0c1a920b76aaed3f8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-19 21:19:29.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-19 21:19:29.000000000","tz":0},"subject":"Skip tempest jobs on api-ref changes","message":"Skip tempest jobs on api-ref changes\n\nChange-Id: Ica9d392be3f12401f7ba51baffb0430bf92ac519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/56383590e74e16f9fdc8ee4b22e9cf3826cf4019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/56383590e74e16f9fdc8ee4b22e9cf3826cf4019"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
