)]}'
{"id":"openstack%2Fzun~662900","triplet_id":"openstack%2Fzun~master~Icfda0a3eb71cc32d16d4c9f9373606bfc08852ce","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"Icfda0a3eb71cc32d16d4c9f9373606bfc08852ce","subject":"Sync Sphinx requirement","status":"MERGED","created":"2019-06-04 01:51:25.000000000","updated":"2019-06-08 15:01:52.000000000","submitted":"2019-06-08 15:01:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662900-1560006112339-edb3e276","meta_rev_id":"6d195c01aedb152b3d1c9e0c5d92e1210179e60d","_number":662900,"virtual_id_number":662900,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-06-08 15:01:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-06-08 14:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-06-05 01:37:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-05 10:44:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-06-08 14:55:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 01:37:21.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-06-05 10:44:57.000000000","updated_by":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"reviewer":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"state":"REVIEWER"},{"updated":"2019-06-08 14:55:21.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2019-06-08 15:01:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efd933e02e09095540e434e760f78f351c913fd0","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-04 01:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e15b6c41d534d054b15426a5081618f7aa9ace9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 02:28:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/00/662900/1/check/build-openstack-api-ref/1146cae/html/ : SUCCESS in 3m 41s\n- requirements-check http://logs.openstack.org/00/662900/1/check/requirements-check/6348a4c/ : SUCCESS in 2m 50s\n- openstack-tox-pep8 http://logs.openstack.org/00/662900/1/check/openstack-tox-pep8/9261f23/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/00/662900/1/check/openstack-tox-docs/7753bfe/html/ : SUCCESS in 3m 26s\n- build-openstack-releasenotes http://logs.openstack.org/00/662900/1/check/build-openstack-releasenotes/579820d/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c4baf1ece7440bb715055673cd69956098d4fc9f","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-05 01:37:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"459b0e8a52cb70c27903c0aa94d9e71aff458cc3","author":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"date":"2019-06-05 10:44:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3b9adb1d9c2cd5d64545f3f325d31430fd1155e5","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-08 14:55:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e798a21b6a9f0ca7f7b6e480c43f25effd4644c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 14:55:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"966a922ec1e5dbcb06c4f9a6f0a71271627f5612","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 15:01:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/00/662900/1/gate/build-openstack-api-ref/a8bd467/html/ : SUCCESS in 4m 17s\n- requirements-check http://logs.openstack.org/00/662900/1/gate/requirements-check/6d3c311/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/00/662900/1/gate/openstack-tox-pep8/b027d9c/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/00/662900/1/gate/openstack-tox-docs/58b4428/html/ : SUCCESS in 3m 52s\n- build-openstack-releasenotes http://logs.openstack.org/00/662900/1/gate/build-openstack-releasenotes/7b90cf7/html/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6d195c01aedb152b3d1c9e0c5d92e1210179e60d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 15:01:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e307cc274ca7fed2b365e2554580211982f84aa","revisions":{"6e307cc274ca7fed2b365e2554580211982f84aa":{"kind":"REWORK","_number":1,"created":"2019-06-04 01:51:25.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/00/662900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/00/662900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/00/662900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/00/662900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/00/662900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/00/662900/1"}}},"commit":{"parents":[{"commit":"83eaf006d6f9749ac85ce14de3a648be6d03011f","subject":"Merge \"Add \u0027init_containers\u0027 field to capsule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/83eaf006d6f9749ac85ce14de3a648be6d03011f"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-06-04 01:49:52.000000000","tz":480},"committer":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-06-04 01:51:21.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\nChange-Id: Icfda0a3eb71cc32d16d4c9f9373606bfc08852ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6e307cc274ca7fed2b365e2554580211982f84aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6e307cc274ca7fed2b365e2554580211982f84aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
