)]}'
{"id":"openstack%2Fzun~664142","triplet_id":"openstack%2Fzun~master~I6d041fef679caf5ac2546ca1a8fdbf31e5322dec","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I6d041fef679caf5ac2546ca1a8fdbf31e5322dec","subject":"Allow force delete a \u0027DELETING\u0027 container","status":"MERGED","created":"2019-06-08 20:26:13.000000000","updated":"2019-07-02 02:29:58.000000000","submitted":"2019-07-02 02:29:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664142-1562034598411-a6540184","meta_rev_id":"76a5bdd7fd134b130b4c852a01b1fbe6bf4ddc1b","_number":664142,"virtual_id_number":664142,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-07-02 02:29:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":2,"date":"2019-07-02 01:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":1,"date":"2019-07-02 01:17:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 01:17:42.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-07-02 02:29:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84e365af75fc66c779102c9c17f1a48b14c34060","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-08 20:26:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e313d02507628c7189d036f76d2679ff427707e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 21:23:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/42/664142/1/check/openstack-tox-cover/919c565/cover/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/664142/1/check/openstack-tox-lower-constraints/e6fc4ed/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/42/664142/1/check/openstack-tox-pep8/312c5e3/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/42/664142/1/check/openstack-tox-py27/ebd5b38/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/42/664142/1/check/openstack-tox-py36/de9073e/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/42/664142/1/check/openstack-tox-py37/1674115/ : SUCCESS in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/42/664142/1/check/openstack-tox-docs/ffa6769/html/ : SUCCESS in 3m 18s\n- zun-tempest-docker-sql http://logs.openstack.org/42/664142/1/check/zun-tempest-docker-sql/de3235c/ : SUCCESS in 49m 33s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/42/664142/1/check/zun-tempest-py3-docker-sql/49accd7/ : SUCCESS in 46m 35s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/42/664142/1/check/zun-tempest-multinode-docker-sql/887eacd/ : SUCCESS in 56m 22s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/42/664142/1/check/kolla-ansible-ubuntu-source-zun/40e0dcc/ : SUCCESS in 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8645089df0c954427af1940c344f35c661e15548","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-07-02 01:17:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2125098cb5e2dfbb6c0a7a549bd886c41efc2c19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 01:17:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"149c4ba7784680b0d52fb73843b30a65fbb1ccfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 02:29:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"76a5bdd7fd134b130b4c852a01b1fbe6bf4ddc1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 02:29:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/42/664142/1/gate/openstack-tox-lower-constraints/bc1fb5e/ : SUCCESS in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/42/664142/1/gate/openstack-tox-pep8/74741eb/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/42/664142/1/gate/openstack-tox-py27/cd7b508/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/42/664142/1/gate/openstack-tox-py36/51b940a/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/42/664142/1/gate/openstack-tox-py37/e295389/ : SUCCESS in 6m 08s\n- openstack-tox-docs http://logs.openstack.org/42/664142/1/gate/openstack-tox-docs/668a20f/html/ : SUCCESS in 3m 48s\n- zun-tempest-docker-sql http://logs.openstack.org/42/664142/1/gate/zun-tempest-docker-sql/f30225c/ : SUCCESS in 53m 54s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/42/664142/1/gate/zun-tempest-py3-docker-sql/80f7f75/ : SUCCESS in 51m 18s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/42/664142/1/gate/zun-tempest-multinode-docker-sql/e88d5d4/ : SUCCESS in 1h 09m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cc1098c13d682f391ddcda6c86ed23ee60a031c","revisions":{"4cc1098c13d682f391ddcda6c86ed23ee60a031c":{"kind":"REWORK","_number":1,"created":"2019-06-08 20:26:13.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/42/664142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/42/664142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/664142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/664142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/42/664142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/42/664142/1"}}},"commit":{"parents":[{"commit":"6e307cc274ca7fed2b365e2554580211982f84aa","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/6e307cc274ca7fed2b365e2554580211982f84aa"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-08 20:25:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-08 20:25:45.000000000","tz":0},"subject":"Allow force delete a \u0027DELETING\u0027 container","message":"Allow force delete a \u0027DELETING\u0027 container\n\nChange-Id: I6d041fef679caf5ac2546ca1a8fdbf31e5322dec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4cc1098c13d682f391ddcda6c86ed23ee60a031c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/4cc1098c13d682f391ddcda6c86ed23ee60a031c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
