)]}'
{"id":"openstack%2Fzun~668241","triplet_id":"openstack%2Fzun~master~Ibfa5cd8a48502c3dec357437fef0094ab8960a77","project":"openstack/zun","branch":"master","topic":"glance_endpoint_type","hashtags":[],"change_id":"Ibfa5cd8a48502c3dec357437fef0094ab8960a77","subject":"Endpoint type for glanceclient","status":"MERGED","created":"2019-06-28 20:07:03.000000000","updated":"2019-06-30 02:39:25.000000000","submitted":"2019-06-30 02:39:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668241-1561862365264-ca750131","meta_rev_id":"5e47878e8633167b9bc9fa13a157b9e2b2bb37e9","_number":668241,"virtual_id_number":668241,"owner":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":2,"date":"2019-06-30 02:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-06-30 01:36:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-06-30 01:36:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 20:07:14.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-06-28 20:15:45.000000000","updated_by":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"reviewer":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"state":"REVIEWER"},{"updated":"2019-06-30 01:36:49.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2019-06-30 02:39:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad6e17ab631c904017987a7921a0849735508588","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2019-06-28 20:07:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90c96a879c715cec702f2dcd8cf1e2d60b5fc885","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-06-28 20:07:14.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"a8b4367f4915ba1d7ce69195c16ff3434ee060c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 21:11:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/41/668241/1/check/openstack-tox-cover/e295b77/cover/ : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/668241/1/check/openstack-tox-lower-constraints/4fcef4c/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/41/668241/1/check/openstack-tox-pep8/188bb6c/ : FAILURE in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/41/668241/1/check/openstack-tox-py27/4821cc2/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/41/668241/1/check/openstack-tox-py36/7c94f50/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/41/668241/1/check/openstack-tox-py37/b2d0770/ : SUCCESS in 4m 31s\n- openstack-tox-docs http://logs.openstack.org/41/668241/1/check/openstack-tox-docs/eba672d/html/ : SUCCESS in 3m 39s\n- zun-tempest-docker-sql http://logs.openstack.org/41/668241/1/check/zun-tempest-docker-sql/3ec9c81/ : SUCCESS in 52m 16s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/41/668241/1/check/zun-tempest-py3-docker-sql/4f1c803/ : SUCCESS in 51m 08s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/41/668241/1/check/zun-tempest-multinode-docker-sql/866f059/ : SUCCESS in 1h 00m 57s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/41/668241/1/check/kolla-ansible-ubuntu-source-zun/35f19ba/ : SUCCESS in 42m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b24110fd40da796610403a3427d920bb2f18b3e","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2019-06-28 21:19:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8c42679ba503cabd2609dffaae0d2c335cc85da5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 22:42:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/41/668241/2/check/openstack-tox-cover/714bffe/cover/ : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/668241/2/check/openstack-tox-lower-constraints/15d7e0c/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/41/668241/2/check/openstack-tox-pep8/193e830/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/41/668241/2/check/openstack-tox-py27/ac95d69/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/41/668241/2/check/openstack-tox-py36/ae3ef5b/ : SUCCESS in 3m 48s\n- openstack-tox-py37 http://logs.openstack.org/41/668241/2/check/openstack-tox-py37/ed25442/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/41/668241/2/check/openstack-tox-docs/5b7ee41/html/ : SUCCESS in 3m 05s\n- zun-tempest-docker-sql http://logs.openstack.org/41/668241/2/check/zun-tempest-docker-sql/6a690e5/ : SUCCESS in 50m 45s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/41/668241/2/check/zun-tempest-py3-docker-sql/d6c790e/ : SUCCESS in 46m 25s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/41/668241/2/check/zun-tempest-multinode-docker-sql/398ab2c/ : SUCCESS in 1h 01m 36s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/41/668241/2/check/kolla-ansible-ubuntu-source-zun/9288f9c/ : SUCCESS in 49m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c90605832bbecebc5e314195e0d7840dfd45c03","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-29 20:23:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4af0f53fd6f17fe4018ce8a0e5dc64ddd4bb576","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-29 23:13:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"72bc6fa541e03856b911729d983f9cf536279bda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 00:17:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/41/668241/3/check/openstack-tox-cover/ac66c1b/cover/ : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/668241/3/check/openstack-tox-lower-constraints/7bed37f/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/41/668241/3/check/openstack-tox-pep8/4a745be/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/41/668241/3/check/openstack-tox-py27/05c63a2/ : SUCCESS in 3m 46s\n- openstack-tox-py36 http://logs.openstack.org/41/668241/3/check/openstack-tox-py36/a3ac2fe/ : SUCCESS in 3m 58s\n- openstack-tox-py37 http://logs.openstack.org/41/668241/3/check/openstack-tox-py37/7efa854/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/41/668241/3/check/openstack-tox-docs/a3590a0/html/ : SUCCESS in 4m 09s\n- zun-tempest-docker-sql http://logs.openstack.org/41/668241/3/check/zun-tempest-docker-sql/fab200b/ : SUCCESS in 47m 12s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/41/668241/3/check/zun-tempest-py3-docker-sql/c8ee1b0/ : SUCCESS in 46m 58s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/41/668241/3/check/zun-tempest-multinode-docker-sql/b68fe14/ : FAILURE in 58m 31s\n- zun-fullstack zun-fullstack : ERROR Unable to find playbook /var/lib/zuul/builds/fb65f3ef6be64081809b6df06a5cc55d/untrusted/project_5/opendev.org/openstack/zun/playbooks/fullstack/run.yaml (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/41/668241/3/check/kolla-ansible-ubuntu-source-zun/04a1c1c/ : SUCCESS in 37m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"273827a4e34924b763944007e2b0955609e8f8b2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-30 00:24:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8689bc7ab6108b9e91cfb5ea5fbe20042808b50e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-30 00:25:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ca58af1250cdd1505fb9a0c47dd3cdb2c75a990a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-30 00:29:37.000000000","message":"Patch Set 3:\n\n@Stanislav Dmitriev,\n\nThanks for first contribution on Zun. I made a few modification, mainly for fixing the unit tests. The change looks good to me. Thanks for the patch.","accounts_in_message":[],"_revision_number":3},{"id":"75e60d3dafa899caeec058b8a68e8ab605986f5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 01:29:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/41/668241/3/check/openstack-tox-cover/90e9dd1/cover/ : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/668241/3/check/openstack-tox-lower-constraints/adbd462/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/41/668241/3/check/openstack-tox-pep8/410a179/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/41/668241/3/check/openstack-tox-py27/17dee46/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/41/668241/3/check/openstack-tox-py36/872814d/ : SUCCESS in 4m 01s\n- openstack-tox-py37 http://logs.openstack.org/41/668241/3/check/openstack-tox-py37/59ad1a3/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/41/668241/3/check/openstack-tox-docs/4ea48e0/html/ : SUCCESS in 3m 14s\n- zun-tempest-docker-sql http://logs.openstack.org/41/668241/3/check/zun-tempest-docker-sql/e402caf/ : SUCCESS in 47m 24s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/41/668241/3/check/zun-tempest-py3-docker-sql/fecd7f6/ : SUCCESS in 42m 19s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/41/668241/3/check/zun-tempest-multinode-docker-sql/19f9654/ : SUCCESS in 59m 37s\n- zun-fullstack http://logs.openstack.org/41/668241/3/check/zun-fullstack/7a66f44/ : SUCCESS in 33m 38s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/41/668241/3/check/kolla-ansible-ubuntu-source-zun/167ece0/ : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ab74f4f9dc3194c990121bd87ec501e5f48961f4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-30 01:36:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b1f767122070890199e07e7c6687d54ac312bd90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 01:37:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f9061029b722cd587ac97ba93c8d8ee7cb29c81d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 02:39:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/41/668241/3/gate/openstack-tox-lower-constraints/9474642/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/41/668241/3/gate/openstack-tox-pep8/fef4fd6/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/41/668241/3/gate/openstack-tox-py27/1f40f20/ : SUCCESS in 6m 03s\n- openstack-tox-py36 http://logs.openstack.org/41/668241/3/gate/openstack-tox-py36/be12e6f/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/41/668241/3/gate/openstack-tox-py37/9dbeec8/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/41/668241/3/gate/openstack-tox-docs/b1b8b2e/html/ : SUCCESS in 3m 00s\n- zun-tempest-docker-sql http://logs.openstack.org/41/668241/3/gate/zun-tempest-docker-sql/04b63d2/ : SUCCESS in 51m 55s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/41/668241/3/gate/zun-tempest-py3-docker-sql/ef9780d/ : SUCCESS in 44m 41s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/41/668241/3/gate/zun-tempest-multinode-docker-sql/1815a0c/ : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":3},{"id":"aaca2d529cd5681f1e77f2f19c4ef5f793188409","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 02:39:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bcd102c97e7aa6cff77273aa487c7dd70704d8d0","revisions":{"57bd53cac86f501120b0abebb4766dc3b2cb5779":{"kind":"REWORK","_number":1,"created":"2019-06-28 20:07:03.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/41/668241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/41/668241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/41/668241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/41/668241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/41/668241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/41/668241/1"}}},"commit":{"parents":[{"commit":"009424db69da469a31165858816f6fc8c30c2921","subject":"Switch to the new canonical constraints URL on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/009424db69da469a31165858816f6fc8c30c2921"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2019-06-28 19:45:21.000000000","tz":0},"committer":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2019-06-28 19:45:21.000000000","tz":0},"subject":"Endpoint type for glanceclient","message":"Endpoint type for glanceclient\n\n    At this moment nevertheless of \u0027endpoint_type\u0027 parameter from\n\u0027glance_client\u0027 section Zun always tries to interact with Glance\npublic API endpoint. If public API endpoint unaccessible from\ncompute nodes zun-compute service is unable to run new container\nfrom image stored in the glance repository.\n    This patch intended to make Zun use \u0027endpoint_type\u0027 parameter\nfrom \u0027glance_client\u0027 section of zun.conf file and pass the same\ntype of endpoint to the glanceclient.\n\nChange-Id: Ibfa5cd8a48502c3dec357437fef0094ab8960a77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/57bd53cac86f501120b0abebb4766dc3b2cb5779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/57bd53cac86f501120b0abebb4766dc3b2cb5779"}]},"branch":"refs/heads/master"},"5f4262d5766e5394be165a95361fba9e86f9e6b5":{"kind":"REWORK","_number":2,"created":"2019-06-28 21:19:13.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/41/668241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/41/668241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/41/668241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/41/668241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/41/668241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/41/668241/2"}}},"commit":{"parents":[{"commit":"009424db69da469a31165858816f6fc8c30c2921","subject":"Switch to the new canonical constraints URL on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/009424db69da469a31165858816f6fc8c30c2921"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2019-06-28 19:45:21.000000000","tz":0},"committer":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2019-06-28 21:18:38.000000000","tz":0},"subject":"Endpoint type for glanceclient","message":"Endpoint type for glanceclient\n\n    At this moment nevertheless of \u0027endpoint_type\u0027 parameter from\n\u0027glance_client\u0027 section Zun always tries to interact with Glance\npublic API endpoint. If public API endpoint unaccessible from\ncompute nodes zun-compute service is unable to run new container\nfrom image stored in the glance repository.\n    This patch intended to make Zun use \u0027endpoint_type\u0027 parameter\nfrom \u0027glance_client\u0027 section of zun.conf file and pass the same\ntype of endpoint to the glanceclient.\n\nChange-Id: Ibfa5cd8a48502c3dec357437fef0094ab8960a77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5f4262d5766e5394be165a95361fba9e86f9e6b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/5f4262d5766e5394be165a95361fba9e86f9e6b5"}]},"branch":"refs/heads/master"},"bcd102c97e7aa6cff77273aa487c7dd70704d8d0":{"kind":"REWORK","_number":3,"created":"2019-06-29 23:13:29.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/41/668241/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/41/668241/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/41/668241/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/41/668241/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/41/668241/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/41/668241/3"}}},"commit":{"parents":[{"commit":"009424db69da469a31165858816f6fc8c30c2921","subject":"Switch to the new canonical constraints URL on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/009424db69da469a31165858816f6fc8c30c2921"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2019-06-28 19:45:21.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-29 23:13:28.000000000","tz":0},"subject":"Endpoint type for glanceclient","message":"Endpoint type for glanceclient\n\nAt this moment nevertheless of \u0027endpoint_type\u0027 parameter from\n\u0027glance_client\u0027 section Zun always tries to interact with Glance\npublic API endpoint. If public API endpoint unaccessible from\ncompute nodes zun-compute service is unable to run new container\nfrom image stored in the glance repository.\n\nThis patch intended to make Zun use \u0027endpoint_type\u0027 parameter\nfrom \u0027glance_client\u0027 section of zun.conf file and pass the same\ntype of endpoint to the glanceclient.\n\nChange-Id: Ibfa5cd8a48502c3dec357437fef0094ab8960a77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/bcd102c97e7aa6cff77273aa487c7dd70704d8d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/bcd102c97e7aa6cff77273aa487c7dd70704d8d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
