)]}'
{"id":"openstack%2Fzun~669554","triplet_id":"openstack%2Fzun~stable%2Frocky~I7fff1f20d50eacc4eb2f0b0f6f24ba521f5df5f7","project":"openstack/zun","branch":"stable/rocky","hashtags":[],"change_id":"I7fff1f20d50eacc4eb2f0b0f6f24ba521f5df5f7","subject":"List system dependencies for running common tests","status":"MERGED","created":"2019-07-06 20:19:40.000000000","updated":"2019-07-06 22:21:35.000000000","submitted":"2019-07-06 22:21:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669554-1562451695701-29425bd4","meta_rev_id":"dc3bab077c5aebea93245c7f12a90b756f1c8d48","_number":669554,"virtual_id_number":669554,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-07-06 22:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-07-06 21:20:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-07-06 21:20:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-06 22:21:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c7d237d5b91f97b81cb2fcbaab408f225505f82","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-07-06 20:19:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"589354dcabfa8030265508ccdf4a5bcfb3ba6d85","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-07-06 21:20:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"58e1f6f3e34ebe875f16fad78daa2506ad589618","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 21:28:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/54/669554/1/check/openstack-tox-pep8/0de3eb8/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/54/669554/1/check/openstack-tox-py27/4dbd025/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/54/669554/1/check/openstack-tox-py35/28ee99e/ : SUCCESS in 3m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/54/669554/1/check/build-openstack-sphinx-docs/ea515eb/html/ : SUCCESS in 3m 30s\n- zun-tempest-docker-sql http://logs.openstack.org/54/669554/1/check/zun-tempest-docker-sql/e65108f/ : SUCCESS in 45m 54s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/54/669554/1/check/zun-tempest-py35-docker-sql/8290612/ : SUCCESS in 45m 00s\n- zun-tempest-docker-etcd http://logs.openstack.org/54/669554/1/check/zun-tempest-docker-etcd/cdfaf6a/ : SUCCESS in 45m 50s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/54/669554/1/check/zun-tempest-multinode-docker-sql/1a9eaee/ : FAILURE in 1h 08m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/54/669554/1/check/openstack-tox-lower-constraints/e20a103/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a7231b6e7eb3939dc2c40492a98dfdcdbfa7560a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 21:29:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0701f17fd04ead50893f1b07d2b435275ba318f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 22:21:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dc3bab077c5aebea93245c7f12a90b756f1c8d48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 22:21:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/54/669554/1/gate/openstack-tox-pep8/33a2b7c/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/54/669554/1/gate/openstack-tox-py27/7198692/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/54/669554/1/gate/openstack-tox-py35/4f2b1a7/ : SUCCESS in 4m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/54/669554/1/gate/build-openstack-sphinx-docs/c556693/html/ : SUCCESS in 4m 15s\n- zun-tempest-docker-sql http://logs.openstack.org/54/669554/1/gate/zun-tempest-docker-sql/1339d46/ : SUCCESS in 47m 25s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/54/669554/1/gate/zun-tempest-py35-docker-sql/47f3885/ : SUCCESS in 42m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/669554/1/gate/openstack-tox-lower-constraints/46395d4/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70e684eb4fa962d1da4c9c69a055cfce11179ae2","revisions":{"70e684eb4fa962d1da4c9c69a055cfce11179ae2":{"kind":"REWORK","_number":1,"created":"2019-07-06 20:19:40.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/54/669554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/54/669554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/54/669554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/54/669554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/54/669554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/54/669554/1"}}},"commit":{"parents":[{"commit":"8961b948c3fc842dc8a582c5d8f2faa4d04e15c4","subject":"Fix requirement job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/8961b948c3fc842dc8a582c5d8f2faa4d04e15c4"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2018-10-13 20:45:54.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-07-06 20:19:40.000000000","tz":0},"subject":"List system dependencies for running common tests","message":"List system dependencies for running common tests\n\nAdd an bindep.txt file containing a cross-platform list of\ndependencies needed for running included tox-based tests. Also include a\ntox environment for convenience calling the bindep[*] utility to list\nany missing system requirements.\n\nFor bindep.txt see also\nhttp://docs.openstack.org/infra/manual/drivers.html#package-requirements\n\n[*] http://docs.openstack.org/infra/bindep/\n\nChange-Id: I7fff1f20d50eacc4eb2f0b0f6f24ba521f5df5f7\n(cherry picked from commit 9a13755cbe80f8eff88b8728b7250d8418cd768e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/70e684eb4fa962d1da4c9c69a055cfce11179ae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/70e684eb4fa962d1da4c9c69a055cfce11179ae2"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
