)]}'
{"id":"openstack%2Fzun~670221","triplet_id":"openstack%2Fzun~master~I1bad75ca91032508147e378ddc8567b41e7b0a88","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I1bad75ca91032508147e378ddc8567b41e7b0a88","subject":"[WIP] No additional property for capsule spec","status":"NEW","created":"2019-07-11 03:08:08.000000000","updated":"2019-07-11 04:07:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"180b92b3c65110e773b0ca94ce2e7aa41a2ff6c3","_number":670221,"virtual_id_number":670221,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":-1,"date":"2019-07-11 04:07:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":-1,"date":"2019-07-11 03:08:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 04:07:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b0e00624a3037ff8335f36f82792baf6c5cf06c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-07-11 03:08:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94be52241e48a354e4a89ab3bbae23656d7be8dd","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-07-11 03:08:23.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"180b92b3c65110e773b0ca94ce2e7aa41a2ff6c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 04:07:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/21/670221/1/check/openstack-tox-cover/fc8c3eb/ : FAILURE in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670221/1/check/openstack-tox-lower-constraints/31be1e3/ : FAILURE in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/21/670221/1/check/openstack-tox-pep8/dfbcec6/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/21/670221/1/check/openstack-tox-py27/a00e38b/ : FAILURE in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/21/670221/1/check/openstack-tox-py36/afd8e2e/ : FAILURE in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/21/670221/1/check/openstack-tox-py37/bf151f7/ : FAILURE in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/21/670221/1/check/openstack-tox-docs/8104bc3/html/ : SUCCESS in 3m 39s\n- zun-tempest-docker-sql http://logs.openstack.org/21/670221/1/check/zun-tempest-docker-sql/5b8ce52/ : SUCCESS in 43m 19s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/21/670221/1/check/zun-tempest-py3-docker-sql/d74d4d8/ : SUCCESS in 42m 54s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/21/670221/1/check/zun-tempest-multinode-docker-sql/4d8ad8c/ : SUCCESS in 55m 16s\n- zun-fullstack http://logs.openstack.org/21/670221/1/check/zun-fullstack/6ba8102/ : SUCCESS in 32m 13s (non-voting)\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/21/670221/1/check/kolla-ansible-ubuntu-source-zun/3e9815c/ : SUCCESS in 36m 24s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb7a940df4b2b41e622e2a4405e4dfaedc1589f5","revisions":{"eb7a940df4b2b41e622e2a4405e4dfaedc1589f5":{"kind":"REWORK","_number":1,"created":"2019-07-11 03:08:08.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/21/670221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/21/670221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/670221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/670221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/21/670221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/21/670221/1"}}},"commit":{"parents":[{"commit":"23b3b7cee1a156f3048d744f5f4eac594e9e3aa1","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/23b3b7cee1a156f3048d744f5f4eac594e9e3aa1"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-07-11 03:07:36.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-07-11 03:07:36.000000000","tz":0},"subject":"[WIP] No additional property for capsule spec","message":"[WIP] No additional property for capsule spec\n\nChange-Id: I1bad75ca91032508147e378ddc8567b41e7b0a88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/eb7a940df4b2b41e622e2a4405e4dfaedc1589f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/eb7a940df4b2b41e622e2a4405e4dfaedc1589f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
