)]}'
{"id":"openstack%2Fzun~675786","triplet_id":"openstack%2Fzun~master~I4a2b2a24f75757d1fb9cd925eecf386b2a1752eb","project":"openstack/zun","branch":"master","hashtags":[],"change_id":"I4a2b2a24f75757d1fb9cd925eecf386b2a1752eb","subject":"Remove the unnecessary glance_store config","status":"MERGED","created":"2019-08-11 16:23:45.000000000","updated":"2019-08-12 02:34:09.000000000","submitted":"2019-08-12 02:34:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675786-1565577249693-674d1078","meta_rev_id":"831d275416995f5926c0fbd2a3dc44fae140c891","_number":675786,"virtual_id_number":675786,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":2,"date":"2019-08-12 02:34:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":2,"date":"2019-08-12 01:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"all":[{"value":1,"date":"2019-08-12 01:21:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 01:21:55.000000000","updated_by":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"reviewer":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"state":"REVIEWER"},{"updated":"2019-08-12 02:34:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd52248af4b943cef32618a79bab6f26085f6b9a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-11 16:23:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1bd665a44e270a4abee594867a84347a00b4766","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-11 17:36:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/86/675786/1/check/openstack-tox-cover/6da06f2/cover/ : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://logs.opendev.org/86/675786/1/check/openstack-tox-lower-constraints/941ce53/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://logs.opendev.org/86/675786/1/check/openstack-tox-pep8/bf80b94/ : SUCCESS in 4m 04s\n- openstack-tox-py27 https://logs.opendev.org/86/675786/1/check/openstack-tox-py27/55371f4/ : SUCCESS in 4m 15s\n- openstack-tox-py36 https://logs.opendev.org/86/675786/1/check/openstack-tox-py36/c1822c5/ : SUCCESS in 5m 44s\n- openstack-tox-py37 https://logs.opendev.org/86/675786/1/check/openstack-tox-py37/bc3536f/ : SUCCESS in 5m 27s\n- openstack-tox-docs https://logs.opendev.org/86/675786/1/check/openstack-tox-docs/2fa0d31/html/ : SUCCESS in 3m 37s\n- zun-tempest-docker-sql https://logs.opendev.org/86/675786/1/check/zun-tempest-docker-sql/d4e7409/ : SUCCESS in 49m 35s\n- zun-tempest-py3-docker-sql https://logs.opendev.org/86/675786/1/check/zun-tempest-py3-docker-sql/6be87dd/ : SUCCESS in 41m 50s\n- zun-tempest-multinode-docker-sql https://logs.opendev.org/86/675786/1/check/zun-tempest-multinode-docker-sql/6fd52fa/ : SUCCESS in 1h 08m 09s\n- zun-fullstack https://logs.opendev.org/86/675786/1/check/zun-fullstack/42f80ff/ : SUCCESS in 26m 22s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://logs.opendev.org/86/675786/1/check/kolla-ansible-ubuntu-source-zun/e49205d/ : SUCCESS in 46m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0260bdc935439a8efdf4232a46798ebcb0ac940","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-08-12 01:21:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c347157e2ac80f4e73d9af4368820de5c42b881c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 01:22:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80342f7dbc8bf60b7a56573e5c92aef47a4df06a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 02:34:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"831d275416995f5926c0fbd2a3dc44fae140c891","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 02:34:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/86/675786/1/gate/openstack-tox-lower-constraints/73106df/ : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://logs.opendev.org/86/675786/1/gate/openstack-tox-pep8/775a454/ : SUCCESS in 4m 17s\n- openstack-tox-py27 https://logs.opendev.org/86/675786/1/gate/openstack-tox-py27/d91c7da/ : SUCCESS in 4m 21s\n- openstack-tox-py36 https://logs.opendev.org/86/675786/1/gate/openstack-tox-py36/4af6baf/ : SUCCESS in 4m 39s\n- openstack-tox-py37 https://logs.opendev.org/86/675786/1/gate/openstack-tox-py37/984ef77/ : SUCCESS in 5m 02s\n- openstack-tox-docs https://logs.opendev.org/86/675786/1/gate/openstack-tox-docs/8b3ae9e/html/ : SUCCESS in 3m 32s\n- zun-tempest-docker-sql https://logs.opendev.org/86/675786/1/gate/zun-tempest-docker-sql/5370803/ : SUCCESS in 45m 35s\n- zun-tempest-py3-docker-sql https://logs.opendev.org/86/675786/1/gate/zun-tempest-py3-docker-sql/d3b0963/ : SUCCESS in 45m 39s\n- zun-tempest-multinode-docker-sql https://logs.opendev.org/86/675786/1/gate/zun-tempest-multinode-docker-sql/3174b03/ : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67925c9ce93e8d958893ea846aab2051d40791c3","revisions":{"67925c9ce93e8d958893ea846aab2051d40791c3":{"kind":"REWORK","_number":1,"created":"2019-08-11 16:23:45.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/86/675786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/86/675786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/86/675786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/86/675786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/86/675786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/86/675786/1"}}},"commit":{"parents":[{"commit":"bacbaaa0e2a755e0571f7c856296a671860c64f8","subject":"Merge \"Fix format issue on private registry doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/bacbaaa0e2a755e0571f7c856296a671860c64f8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-08-11 16:17:28.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-08-11 16:17:28.000000000","tz":0},"subject":"Remove the unnecessary glance_store config","message":"Remove the unnecessary glance_store config\n\nThe override of the \u0027default_store\u0027 config is a temporary walk-around\na python3 issue. Such walk-around is not needed anymore.\n\nChange-Id: I4a2b2a24f75757d1fb9cd925eecf386b2a1752eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/67925c9ce93e8d958893ea846aab2051d40791c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/67925c9ce93e8d958893ea846aab2051d40791c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
