)]}'
{"id":"openstack%2Fzun~677078","triplet_id":"openstack%2Fzun~stable%2Fstein~I9e06054398b53aa5d51b7db8b031322ba38b782a","project":"openstack/zun","branch":"stable/stein","hashtags":[],"change_id":"I9e06054398b53aa5d51b7db8b031322ba38b782a","subject":"Increase exec timeout to 15 seconds","status":"ABANDONED","created":"2019-08-18 02:07:13.000000000","updated":"2019-08-18 04:14:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"59fc06eab8a0ec3aec304dc3e3c0f7ee123f4300","_number":677078,"virtual_id_number":677078,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-08-18 03:09:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":-1,"date":"2019-08-18 02:16:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-18 03:09:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66d4d3889b5200a1bc19ca6962e767480e8fc10e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-18 02:07:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e1f660656340dce71b769561aeadaf8ef055c26","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-18 02:16:24.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7fae8c8cd71152e68f21da884da9ebf874638a04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 03:09:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c66360fe0b04e7a990761d13f59403a : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fabcc413fbb4853821cd5a4596f1018 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/004bb879c0e74844b37aeebc8833bd04 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50052526783f453886e51932a4c08273 : SUCCESS in 6m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f3f6ea39e74c48129ecd6812c8b1d6d5 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff2a01c4a92348249b43b10b3bd6c339 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f69180b2864d4dcab41c93144e6853d7 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba563af68a5d4d6cb95cff2c869009e9 : SUCCESS in 4m 41s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/931e2fedabbe48418478132b9cf8a353 : SUCCESS in 40m 14s\n- zun-tempest-py35-docker-sql https://zuul.opendev.org/t/openstack/build/e18b0f862e454dd98e6a7da9ff0c905e : SUCCESS in 41m 30s\n- zun-tempest-docker-etcd https://zuul.opendev.org/t/openstack/build/93a912c9dc1b4234bdd6af478810945d : SUCCESS in 45m 04s (non-voting)\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/08530da31d6549d6ac35194e3306b441 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":1},{"id":"59fc06eab8a0ec3aec304dc3e3c0f7ee123f4300","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-18 04:14:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc4754d075378a5ef59706e9fea7076c86624538","revisions":{"dc4754d075378a5ef59706e9fea7076c86624538":{"kind":"REWORK","_number":1,"created":"2019-08-18 02:07:13.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/78/677078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun","ref":"refs/changes/78/677078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun refs/changes/78/677078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun refs/changes/78/677078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun refs/changes/78/677078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun refs/changes/78/677078/1"}}},"commit":{"parents":[{"commit":"28fb8324b7a0feb88899d5d855519f03dbab4e9f","subject":"Endpoint type for glanceclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/28fb8324b7a0feb88899d5d855519f03dbab4e9f"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-08-18 02:06:52.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-08-18 02:06:52.000000000","tz":0},"subject":"Increase exec timeout to 15 seconds","message":"Increase exec timeout to 15 seconds\n\nChange-Id: I9e06054398b53aa5d51b7db8b031322ba38b782a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/dc4754d075378a5ef59706e9fea7076c86624538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun/commit/dc4754d075378a5ef59706e9fea7076c86624538"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
